)]}'
{"id":"openstack%2Fpuppet-aodh~823508","triplet_id":"openstack%2Fpuppet-aodh~master~I8f3201ce798e6a269282fb0e54faed5a0030d1ca","project":"openstack/puppet-aodh","branch":"master","topic":"modern-facts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-30 15:38:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8f3201ce798e6a269282fb0e54faed5a0030d1ca","subject":"Replace usage of legacy facts","status":"ABANDONED","created":"2022-01-05 12:49:43.000000000","updated":"2022-07-30 15:38:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"61d2b514a9bb39fd69fdcced7694c795e9928ae6","_number":823508,"virtual_id_number":823508,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-27 09:46:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2022-03-27 08:39:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 14:32:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d1faa4078a473a5fc34318be192210be60b3408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-05 12:49:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d57eebe631c70c40f4ee5b7d2fb87e23f018332d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-05 12:59:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3492b4524b0a1093d232e8018f0b3072293422d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-05 12:59:54.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"07a64e368140257d132fdcfcaf3591da7f5fbbf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 14:32:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/334a8954a6a34a04baa7766e5638ef51 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dd6a668aa7643078556961a83232868 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0a68fbaad6e479880d9ae6ee7d7378b : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d816d0c16b94d45b98f81aa45241553 : FAILURE in 12m 48s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bc601e27ea04be98193f92888fdec60 : FAILURE in 12m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0318d2ad748474e9c3a0ff5f7920cf6 : FAILURE in 9m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fdc9a0e6daa48d8ad769b4bc17c4071 : FAILURE in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac1a2893a17426ab991a6e537aea87d : FAILURE in 11m 02s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9760403942e44a54a35c77690931b803 : FAILURE in 34m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5de7cb3c61b405bb9418083c4d62e5a : SUCCESS in 1h 23m 59s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a745b26b1d242599946474dc5c733d4 : SUCCESS in 1h 00m 12s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccf2d886725a428eb8a7ad52ede6d5c4 : SUCCESS in 15m 59s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/afd267759cac4d8a81f6ee71cde3384e : SUCCESS in 22m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a84f4b179d2486eb4edbcff3e69d040 : SUCCESS in 16m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37faef04f05a4fb9aefafd09429c47d2 : SUCCESS in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15166400c7fb8a90dc7d5d640d67d15bb1e3ca2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-27 08:38:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24d8af1aafdaf4b0a3224f1497b1cbfe1d3e96c6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-27 08:39:04.000000000","message":"Patch Set 3: Code-Review-2","accounts_in_message":[],"_revision_number":3},{"id":"a6a7cd9914457702f04139ac7ee132c2c1959056","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-27 08:39:11.000000000","message":"Topic set to modern-facts","accounts_in_message":[],"_revision_number":3},{"id":"d3e5797167c2b6530dce46dbd5b18c284e017557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-27 09:46:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5503d5adc2c842528ff611f26bd23f2d : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf4922662e52463a958be5530847aa4c : SUCCESS in 4m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/569b3d330a6f45f58eb227cc0e8ede45 : SUCCESS in 5m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e8dc794d6bc4507b3e50f16e454ba70 : FAILURE in 11m 14s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/21faace0b93c4cf0b9cffe87e9d2669f : FAILURE in 11m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dccc056c16794a4d84805cafcd9705c3 : FAILURE in 8m 51s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51456d1cecf24c889fa88b70a8a15f73 : FAILURE in 8m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89d9933fe23446358880c2b2276eb684 : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd885a7c64114452bab010f0002845b1 : FAILURE in 31m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bda39b760f4347aca3bd867357394025 : SUCCESS in 59m 04s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f12f8ddd06c04a81bd4f98a85bb4203d : SUCCESS in 45m 47s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/641cb9c03bc94ee0920c8c78af935e59 : FAILURE in 39m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b7e21209cad4c598bc689d136fe8ff7 : FAILURE in 1h 05m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aa8da3c40b54de88cb01b52d2f664e5 : SUCCESS in 52m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ecb48bf9a174b979d2c6315bea34423 : SUCCESS in 16m 06s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9eb7d7fbc7254857b76cdc91a5055fb5 : SUCCESS in 20m 21s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f743488899554231a3b5e2d26e32466f : FAILURE in 8m 53s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ae6abd9b4ea45c9b03311f6693d2f7d : SUCCESS in 15m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f94c25443fe4f528eee1bf3c8cceb33 : SUCCESS in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61d2b514a9bb39fd69fdcced7694c795e9928ae6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 15:38:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8f7842cf857377a4fdba8edd239a7cc52743cbf6","revisions":{"34633344a5e1f440a04d5d21061400d65c6dc8b7":{"kind":"REWORK","_number":1,"created":"2022-01-05 12:49:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/823508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/08/823508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/1"}}},"commit":{"parents":[{"commit":"47c27146c8f7807cd438d4712a1aeb71cb1cac67","subject":"Remove usage of custom os_package_type fact","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/47c27146c8f7807cd438d4712a1aeb71cb1cac67"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-05 12:44:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-05 12:49:35.000000000","tz":540},"subject":"Replace usage of legacy facts","message":"Replace usage of legacy facts\n\nChange-Id: I8f3201ce798e6a269282fb0e54faed5a0030d1ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/34633344a5e1f440a04d5d21061400d65c6dc8b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/34633344a5e1f440a04d5d21061400d65c6dc8b7"}]},"branch":"refs/heads/master"},"1d2395602261290d954dc3dfd5c7ae2f514f84e8":{"kind":"REWORK","_number":2,"created":"2022-01-05 12:59:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/823508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/08/823508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/2"}}},"commit":{"parents":[{"commit":"47c27146c8f7807cd438d4712a1aeb71cb1cac67","subject":"Remove usage of custom os_package_type fact","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/47c27146c8f7807cd438d4712a1aeb71cb1cac67"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-05 12:44:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-05 12:59:32.000000000","tz":540},"subject":"Replace usage of legacy facts","message":"Replace usage of legacy facts\n\nChange-Id: I8f3201ce798e6a269282fb0e54faed5a0030d1ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1d2395602261290d954dc3dfd5c7ae2f514f84e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1d2395602261290d954dc3dfd5c7ae2f514f84e8"}]},"branch":"refs/heads/master"},"8f7842cf857377a4fdba8edd239a7cc52743cbf6":{"kind":"REWORK","_number":3,"created":"2022-03-27 08:38:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/823508/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/08/823508/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/08/823508/3"}}},"commit":{"parents":[{"commit":"38c5e93a3fb13fbd3d51aca2f315d3f514477f4c","subject":"Prepare Yoga RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/38c5e93a3fb13fbd3d51aca2f315d3f514477f4c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-05 12:45:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-27 08:37:17.000000000","tz":540},"subject":"Replace usage of legacy facts","message":"Replace usage of legacy facts\n\n... by modern structured facts. Also, reference to facts are refactored\nby $facts[...] to explicitly force usage of facts, except for the facts\nprovided by openstacklib.\n\nChange-Id: I8f3201ce798e6a269282fb0e54faed5a0030d1ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8f7842cf857377a4fdba8edd239a7cc52743cbf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8f7842cf857377a4fdba8edd239a7cc52743cbf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
