)]}'
{"id":"openstack%2Fpuppet-aodh~825422","triplet_id":"openstack%2Fpuppet-aodh~master~I511a71b3e25e8cd8c2a7cb0b242d961ce8e916fd","project":"openstack/puppet-aodh","branch":"master","topic":"oslo-messaging-ut","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-21 12:09:01.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I511a71b3e25e8cd8c2a7cb0b242d961ce8e916fd","subject":"Avoid testing details of oslo::messaging(::*)","status":"MERGED","created":"2022-01-20 03:35:00.000000000","updated":"2022-01-21 12:42:22.000000000","submitted":"2022-01-21 12:42:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825422-oslo-messaging-ut","meta_rev_id":"638f3a251076d3eaa0f047d2a5f72b0e00774420","_number":825422,"virtual_id_number":825422,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-21 12:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-20 15:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-21 08:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-21 08:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 03:51:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-20 15:57:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-21 08:58:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"99414b38fe3482090e9cec94c353c6fbe6356f72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-20 03:35:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"103253f7867b1775f6e7e59f85da1e921a9d72c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 03:51:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fdadc2ea4b94720a674766650e7cd66 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2b50e8984c24c9e9b5e3e830f016da4 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81e0ddefc4ab48dbbcdad9b523a21bcb : SUCCESS in 12m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3c5337cd842421a9aa68d4bdba98ce4 : FAILURE in 9m 51s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c394b9ed852c48d9bda85c8cafd7bd7d : FAILURE in 14m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/091a0377acc94864ba7883748a9c151e : FAILURE in 14m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1497114d08e44e5b164c86ac7e95006 : FAILURE in 8m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbda019e06fc45d9bb525e1612550723 : FAILURE in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9760b75e129983557dc6ba07983a7dc7c5b0dfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-20 03:57:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4a3c1b5292f4bcdc407eabb7e0a8aba8695d612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 04:12:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d8555180ba44a6b94d6c444e608c475 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc9becfd91f94b299d818d8829476619 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02ba24284a904c93ba0b9b7574ebc4bc : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/449fbcf1db144741bb02e08fafd71f6e : FAILURE in 10m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf37bb7d7d194dd88007d04646699475 : FAILURE in 10m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb13cca82b17494b97b3c33d72017e1b : FAILURE in 7m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/351f924242114db7b96fcc8f7575060f : FAILURE in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aec10fb47cb14f7fa21e4343fcd4ca9c : FAILURE in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59dc26a4e241d1ede0ea38e56a6ef4a90e6373ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-20 07:25:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fcb747690407977d2230f87dabae030a31e80de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-20 07:47:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"345543b11e154683b9e5d0578ad7dff7187a300c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 08:10:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9c4565bd82f440e9748a94a51deeea4 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14fb9bb00bd54f859ee498bfd5e8ac8b : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d5b1820cf854d2eb333d29ae9d207f3 : SUCCESS in 3m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4165b447ceb944a9b40b8a244c5f6ff3 : RETRY_LIMIT in 5m 14s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/49a47a720f5c45eb97455d505a505cbc : RETRY_LIMIT in 9m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adba17a31d6a4a948e86c10c7f08b5f9 : SUCCESS in 8m 17s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/893f7b920ac9409eb32d20a9aa117571 : SUCCESS in 9m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12aee557ee4d450099de7a52dd2bf04e : SUCCESS in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4ca83cb8eb6cf13cce7dab610f091d349ed99f89","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-20 08:38:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bc1ce70e4dfbb6e721745526dfa366e6bc19ce1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 09:06:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38cf577e917349cf9765a8399f3358be : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de31fbca2fd54e81aaa5bc57e0d753ef : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/514b84b7084a417d8fcb9174e93c0344 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6958506903934f0fa4557ee1b65fb8e2 : RETRY_LIMIT in 4m 24s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/93608b244d00439983224cd5efe4fdc6 : RETRY_LIMIT in 8m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5f208d2c85e4e30bce3783d0c7251a6 : SUCCESS in 7m 27s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d2956610a7a4ab58a5bca1ace54252b : SUCCESS in 8m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6f3a6d2dc884e7e9090ba99902e4e91 : SUCCESS in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bf8ce49d8b6c377a950bc723215d7c14dd8d3c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-20 14:30:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a7f4bdba6550bed58cc39c656bc6121531b93f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 14:46:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6049b50a437546ac9704a6d7bd375f8a : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0494352c90864f948d498e76d5e0f34e : SUCCESS in 4m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a79a19e8eb44f3da130441794333c61 : SUCCESS in 3m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea891851d5a6415e8b1e099b26047a09 : SUCCESS in 10m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2f9f1c97a4145d3a1b46c097f728315 : SUCCESS in 10m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed5cc12b9b9f4e5db5c503d093e01c40 : SUCCESS in 8m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/423cfc15dda94c448cda1efc83df249d : SUCCESS in 8m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ab57039636341b4a1ea9969fde36694 : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd8ea5598cb7c60c1f885ce615645d5316ca2c41","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-20 15:57:49.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fd6ea516c6cc51dad37963cb040be6c3da6aff59","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-21 08:58:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8eeabca38c6b8523e765106f47b36c9bd0ea0441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 09:01:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"45b972091c7b8d49ef44ffabf9840d9b142cf4c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 09:16:14.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2bca8b1e69b421b84511b40a944f6f2 : SUCCESS in 8m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b0142d9e1534dfaaacabe5aadb1707b : FAILURE in 8m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a62ed9ad1e6042539acb74a329c45b63 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfbe3e29d2674056bc317a41b45df126 : SUCCESS in 12m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/42fa1df8ff80498989f70f9e386015b5 : SUCCESS in 9m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f3857b98bf3477393bf04deb9e76941 : SUCCESS in 8m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfff88f586264876b137c449320e6d67 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":5},{"id":"f658955a17e7cd225ae5ddacab9067175fca15b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 10:46:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b9718980701f8c62153a7394bee81684737ddf9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 11:04:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b43e1955fc740c0bd93a24525d82331 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22e2143cdf1f43a18579f805c182bb1d : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7adc13a3e39d4efc9d91cecdabb5d550 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d168dc8b6a94e6f94ee9ed605f154b1 : SUCCESS in 12m 03s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f07a00c5f803490aa7506fcc8ee84a67 : SUCCESS in 12m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/520b19d3132a4ed0badd6140b20054a6 : SUCCESS in 10m 14s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7fb4b4ad3694aac9b5411862a77bf55 : SUCCESS in 9m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91c1f0b6ed034495a1b81b6dc9c170e8 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aa1e76b93c074b8cb8bc66c14eee2ed0afba37d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 11:05:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8c277cf6528e0b376c026efdde55fc2a20fe0113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 11:39:03.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ddaea0541484660942d25a9f6cb4752 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9c0f4dc582e4699a8bd58a9e646801a : FAILURE in 6m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbddb552674e4fffb67635cef5377789 : FAILURE in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2793e0be129d432393c2da062dee3f77 : FAILURE in 29m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/65bab5e55b8b4538aed6efd18da5cf41 : SUCCESS in 11m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03bce932b009458781d0b3560cad0905 : SUCCESS in 13m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/346a5af97a4640c29c7cc1906b591a6a : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":5},{"id":"ab10a4710c5c5a6ccd862fd656d20e2bf6fdca49","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 12:09:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a012094d7b44e9304db19301787681696b3ca081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 12:25:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1063fe4cc7ec4857ab8967bb2e9c0428 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69876bca99e74dfe97ea2aea98d8c6a8 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ae25a2512a44204aa7d19420995a120 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdb04c0b1fe048f5b3ce45a371dc5d0c : SUCCESS in 13m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf1b9ba64ab34f05a65e6fef403709a9 : SUCCESS in 12m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4481aadd75f343819926f16f37fe0a88 : SUCCESS in 10m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6da3b3ed78ad4116b25ffb7b59b79742 : SUCCESS in 10m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/848a589438b64ae7803faeebe5635214 : SUCCESS in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"42a4f11d8ffc8ea58788a59eb9ee1e22cb0361aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 12:27:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"02ae964e23481e87bd2de16b145d614a27dfc126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 12:42:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fab7ac7c24b492c87da592d28ed72c4 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e89dd8704d14701aa617f1d149be29c : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a4f045c6c7d40548dc27e11da0f3f5e : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a82e93bde8654675b979e5de06f23334 : SUCCESS in 11m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/189983410e2d43b9ae15b3bb7d41aae5 : SUCCESS in 11m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a0af990d8f74e37ad34bd4540d66c17 : SUCCESS in 8m 14s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0d0b9ecd2664a8897d2b5dacde93f68 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":5},{"id":"638f3a251076d3eaa0f047d2a5f72b0e00774420","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 12:42:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d398887c5119d44e120b7bc6f478ac350a05c625","revisions":{"f126996d053c5d29115dc0beb7e8476d060f960f":{"kind":"REWORK","_number":1,"created":"2022-01-20 03:35:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/825422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/825422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/1"}}},"commit":{"parents":[{"commit":"49df15f709dbdc470534d41de89fc1f46a4440f2","subject":"Merge \"Accept system scope credentials for Keystone API request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/49df15f709dbdc470534d41de89fc1f46a4440f2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 03:34:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 03:34:21.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I511a71b3e25e8cd8c2a7cb0b242d961ce8e916fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f126996d053c5d29115dc0beb7e8476d060f960f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/f126996d053c5d29115dc0beb7e8476d060f960f"}]},"branch":"refs/heads/master"},"c40c9194f754366d8251cab3fcd1c4956f283894":{"kind":"REWORK","_number":2,"created":"2022-01-20 03:57:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/825422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/825422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/2"}}},"commit":{"parents":[{"commit":"49df15f709dbdc470534d41de89fc1f46a4440f2","subject":"Merge \"Accept system scope credentials for Keystone API request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/49df15f709dbdc470534d41de89fc1f46a4440f2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 03:34:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 03:56:52.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I511a71b3e25e8cd8c2a7cb0b242d961ce8e916fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c40c9194f754366d8251cab3fcd1c4956f283894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c40c9194f754366d8251cab3fcd1c4956f283894"}]},"branch":"refs/heads/master"},"063bfa27f8abd48f1a1811ea4855b5823d3c0175":{"kind":"REWORK","_number":3,"created":"2022-01-20 07:25:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/825422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/825422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/3"}}},"commit":{"parents":[{"commit":"49df15f709dbdc470534d41de89fc1f46a4440f2","subject":"Merge \"Accept system scope credentials for Keystone API request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/49df15f709dbdc470534d41de89fc1f46a4440f2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 03:34:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 07:25:40.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I511a71b3e25e8cd8c2a7cb0b242d961ce8e916fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/063bfa27f8abd48f1a1811ea4855b5823d3c0175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/063bfa27f8abd48f1a1811ea4855b5823d3c0175"}]},"branch":"refs/heads/master"},"d510505ac73d3647b8853e85b0b0d3f48377b4a5":{"kind":"REWORK","_number":4,"created":"2022-01-20 07:47:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/825422/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/825422/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/4"}}},"commit":{"parents":[{"commit":"49df15f709dbdc470534d41de89fc1f46a4440f2","subject":"Merge \"Accept system scope credentials for Keystone API request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/49df15f709dbdc470534d41de89fc1f46a4440f2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 03:34:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 07:47:10.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I511a71b3e25e8cd8c2a7cb0b242d961ce8e916fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d510505ac73d3647b8853e85b0b0d3f48377b4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d510505ac73d3647b8853e85b0b0d3f48377b4a5"}]},"branch":"refs/heads/master"},"d398887c5119d44e120b7bc6f478ac350a05c625":{"kind":"REWORK","_number":5,"created":"2022-01-20 14:30:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/825422/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/825422/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/825422/5"}}},"commit":{"parents":[{"commit":"49df15f709dbdc470534d41de89fc1f46a4440f2","subject":"Merge \"Accept system scope credentials for Keystone API request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/49df15f709dbdc470534d41de89fc1f46a4440f2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 03:34:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-20 14:30:26.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: I511a71b3e25e8cd8c2a7cb0b242d961ce8e916fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d398887c5119d44e120b7bc6f478ac350a05c625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d398887c5119d44e120b7bc6f478ac350a05c625"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
