)]}'
{"id":"openstack%2Fpuppet-aodh~826022","triplet_id":"openstack%2Fpuppet-aodh~master~Ib6b5553f0c9b42f8b94e0ff92405466b974aa56d","project":"openstack/puppet-aodh","branch":"master","topic":"service-role","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-24 14:10:45.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ib6b5553f0c9b42f8b94e0ff92405466b974aa56d","subject":"Assign project-scoped service role by default","status":"ABANDONED","created":"2022-01-24 07:56:42.000000000","updated":"2022-08-10 00:48:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e713ca8b33bec2d98aab82de90fbbfe3a0b5b7b6","_number":826022,"virtual_id_number":826022,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-24 15:54:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2022-01-24 08:00:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 09:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b76778aa33aa549c9eb980df01e848afa60becd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 07:56:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88ef6e0208632ae251b8592241fccc4a5bb86566","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 08:00:25.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"acc6ca65639b68a6b6ef889acfb308eb9e80f57e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 09:19:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab1da5b9a5d9456a9d0f5ab57c7c5d67 : SUCCESS in 9m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21175bf9b7324a3a83f683080aff9678 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca7fc08749f54cc5907685688eb62c30 : FAILURE in 7m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1730ef83442f4e99930e41d5f53f328b : SUCCESS in 9m 42s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8335e760b9b44c1d9203ba79950ae218 : SUCCESS in 10m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6b52da175e342cfa43f8886e70a7a84 : SUCCESS in 12m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dae0d61bdb294f16bf883f89a02b5a3a : SUCCESS in 12m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00d4c31465f74547beacb2c0d5aa050a : SUCCESS in 12m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e291a731abc5433d9051a08420ce81ab : FAILURE in 32m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/95850fabf465472ba49094a7b9200d1f : SUCCESS in 1h 17m 52s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8dc38dc80b2439fb8e8f44222c4b3e0 : SUCCESS in 58m 00s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20181c67e09241c88f7b67e72939d434 : SUCCESS in 18m 11s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3e28445b1ab4053ba96e52263f3b026 : SUCCESS in 18m 18s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8933372c050d49d89b0ed169c9c9167e : SUCCESS in 17m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36031351d96f46a58e0508862ad960f1 : SUCCESS in 15m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea1b645734b84b618e494c136aad6330 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f4a30ac6d589ba172a1fdcf0d7e86648a38fb5d3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 14:10:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1ebc204a63a616c846c98dbb98a45506f2d98c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 15:54:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ff2e4df4c804bc587810d13ba9f931e : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7df292f555bc4921a3b0c6eadf12db78 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e0270ce2c414f74a59f9a90048621ad : SUCCESS in 4m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8eef59eb71124f09bf4a1e6cc5f9be33 : SUCCESS in 15m 35s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/01830c3a559b48f5aa65691edb10d856 : SUCCESS in 14m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7b7615ae46f45e2b97d7c536b1e7a62 : SUCCESS in 9m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52777c60439740a0854ce5df1f92627f : SUCCESS in 8m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cb2d8fe174d454fb8b0bd97bd74e9b9 : SUCCESS in 8m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d42608ebce6a45a2ad75b712402b05d1 : FAILURE in 30m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/01fe7941d82145ed99e574a30a374bbf : SUCCESS in 1h 35m 33s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf2d7c1516ab47eaa7786523900cf3ea : SUCCESS in 47m 50s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1498fc570c434c20af524fd5b8bc9a53 : SUCCESS in 15m 08s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/513d2df469774de19d1c3c70a4b9096a : SUCCESS in 20m 09s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d902c8ed391b4670838fa7f53cc6801b : SUCCESS in 12m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7e6fa2afe904e4aa837b626ad04a835 : SUCCESS in 13m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c898d35b744ec88d892ea3c1864961 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e713ca8b33bec2d98aab82de90fbbfe3a0b5b7b6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-10 00:48:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96be515224af8ea4eab9e2ebf947f5eb0b35ac8b","revisions":{"96be515224af8ea4eab9e2ebf947f5eb0b35ac8b":{"kind":"REWORK","_number":1,"created":"2022-01-24 07:56:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/826022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/22/826022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/826022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/826022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/22/826022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/22/826022/1"}}},"commit":{"parents":[{"commit":"d398887c5119d44e120b7bc6f478ac350a05c625","subject":"Avoid testing details of oslo::messaging(::*)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d398887c5119d44e120b7bc6f478ac350a05c625"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-24 07:54:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-24 07:56:33.000000000","tz":540},"subject":"Assign project-scoped service role by default","message":"Assign project-scoped service role by default\n\n... so that token validation is allowed when SRBAC is enforced in\nKeystone.\n\nChange-Id: Ib6b5553f0c9b42f8b94e0ff92405466b974aa56d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/96be515224af8ea4eab9e2ebf947f5eb0b35ac8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/96be515224af8ea4eab9e2ebf947f5eb0b35ac8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
