)]}'
{"id":"openstack%2Fpuppet-aodh~829281","triplet_id":"openstack%2Fpuppet-aodh~master~I9afc6d9b23ddb72d21675b1ffdcce8c4a25477df","project":"openstack/puppet-aodh","branch":"master","topic":"c9-ut","hashtags":[],"change_id":"I9afc6d9b23ddb72d21675b1ffdcce8c4a25477df","subject":"DNM: Test unit tests for CentOS 9","status":"ABANDONED","created":"2022-02-15 14:58:35.000000000","updated":"2022-02-15 16:36:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"23972033ac42e296432449cd84e4646cec5157ae","_number":829281,"virtual_id_number":829281,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-15 16:22:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 16:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"666c35d70f9afa8122fc5a0c1321db3c644ba99d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 14:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"368eeaad32fbc88b24b3e21ad1dd07874cc88df4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:05:52.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ff5525098629fba1746cfa95bf0338c53cf20067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:17:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"71f7070e90e3b57e7725451eae1a8eabf7e6c17b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:22:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f047b8010064e6591744a2aec298981 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c59cdcfc4ab34a0e8c165844399fc2ca : SUCCESS in 4m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa28299970024ad4b436078f01d7f1b7 : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/23608a94daf64703ab80c0ca59cd9b4c : SUCCESS in 13m 27s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/be6921acd71e45e09146285ab6be5c16 : SUCCESS in 14m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d54e21cb8c0d43ba92c7b7ccd9e6173f : SUCCESS in 10m 29s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5aa5079274ed4e26987652739f9af089 : SUCCESS in 12m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcb9a861b3bb4bdeb48a237a6e31d7f7 : SUCCESS in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23972033ac42e296432449cd84e4646cec5157ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 16:36:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9149e03347b1951345bfc97beb743d2dd95d2e8c","revisions":{"7d8c72945cc9514685dbea45a65a2cd4ce2af56a":{"kind":"REWORK","_number":1,"created":"2022-02-15 14:58:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/829281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/81/829281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/1"}}},"commit":{"parents":[{"commit":"fd7caedd5f320503e893306ad3f33a3b565b6ec4","subject":"Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fd7caedd5f320503e893306ad3f33a3b565b6ec4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:58:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:58:03.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\nChange-Id: I9afc6d9b23ddb72d21675b1ffdcce8c4a25477df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7d8c72945cc9514685dbea45a65a2cd4ce2af56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7d8c72945cc9514685dbea45a65a2cd4ce2af56a"}]},"branch":"refs/heads/master"},"9d1a9f075e6b3cc0836553713343c74c6d6a13c9":{"kind":"NO_CHANGE","_number":2,"created":"2022-02-15 15:05:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/829281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/81/829281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/2"}}},"commit":{"parents":[{"commit":"3c9a5d27cc890e0b847f6c97285a524281f7c4bc","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3c9a5d27cc890e0b847f6c97285a524281f7c4bc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:58:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:05:45.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\nChange-Id: I9afc6d9b23ddb72d21675b1ffdcce8c4a25477df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9d1a9f075e6b3cc0836553713343c74c6d6a13c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9d1a9f075e6b3cc0836553713343c74c6d6a13c9"}]},"branch":"refs/heads/master"},"9149e03347b1951345bfc97beb743d2dd95d2e8c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-15 15:17:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/829281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/81/829281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/81/829281/3"}}},"commit":{"parents":[{"commit":"0f1d26cdb67f4561be291db65cd4574ac0615cb5","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0f1d26cdb67f4561be291db65cd4574ac0615cb5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:58:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:17:08.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\nChange-Id: I9afc6d9b23ddb72d21675b1ffdcce8c4a25477df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9149e03347b1951345bfc97beb743d2dd95d2e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9149e03347b1951345bfc97beb743d2dd95d2e8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
