)]}'
{"id":"openstack%2Fpuppet-aodh~833702","triplet_id":"openstack%2Fpuppet-aodh~master~Iaa53db36697059121868f6415c7e55d0409c50f3","project":"openstack/puppet-aodh","branch":"master","topic":"bug/1964824","hashtags":[],"change_id":"Iaa53db36697059121868f6415c7e55d0409c50f3","subject":"Add $memcached_ipv6 in keystone/authtoken.pp","status":"ABANDONED","created":"2022-03-15 01:35:10.000000000","updated":"2022-03-26 16:36:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"2fd9ce489ef2f4a8b17502763f2ab565d8d3d6e7","_number":833702,"virtual_id_number":833702,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-15 03:29:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-03-16 19:53:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-03-15 03:29:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-16 19:53:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a5d22e75d921ee1aff7542bf77602f716dbcca63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-03-15 01:35:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f03fd956932c64200036677b5e49338ccde38cb2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-03-15 01:41:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"6fb1fc10fbb3f8281f25732e01bd63d10b94bd4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 03:29:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69c5fd547d664e07a936cda031bd1441 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e1c0913b5954739ab00134d4b779a8b : SUCCESS in 5m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ebb2b3bb5b4461c97adaa71f11390f9 : SUCCESS in 5m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0c2eef4cad14d07b24c3908c3e52da4 : SUCCESS in 14m 55s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5cf47852b724f3cae8b9e28390f7428 : SUCCESS in 15m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bba585fbfcbd4f108b782e3acd0452d3 : SUCCESS in 12m 33s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84049f0b1c25494595f1c9e9d697ca87 : SUCCESS in 12m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac6ac26b59a94cb9bf923519f6183e98 : SUCCESS in 12m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/106dc6177f3e4eddb46025c6b31f19ae : FAILURE in 35m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/02ba10ce51184ba7a141ec1065205fec : SUCCESS in 1h 05m 34s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51a1df285fa245f9a67cb76f90fa00da : SUCCESS in 58m 28s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c32742c79e684cf6a6bf82e5a314f5eb : FAILURE in 44m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db5d78329c8844edaa196139c3fc6ad4 : SUCCESS in 1h 12m 51s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c61d827c98834b78ba0d0935f36497f6 : SUCCESS in 56m 10s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/825c7648e4da4cad993143469114c0e2 : SUCCESS in 15m 54s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/438bf13e02ef403593d8361ab2191e5a : SUCCESS in 23m 30s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/686579fe1629430c923189fb73137ed6 : FAILURE in 12m 45s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab81aa4a6697400385dd8ddafe7cddc8 : SUCCESS in 16m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99ae21c4558a45259457148630506c4f : SUCCESS in 18m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f66e8ff9ddf74492b72d769fa38b4586 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"caf738293ee79ee8c8d5a9f75e7af3611ed4295a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-16 19:53:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2fd9ce489ef2f4a8b17502763f2ab565d8d3d6e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-03-26 16:36:08.000000000","message":"Abandoned\n\nFixed elsewhere.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c24e3bb4dfa85b51fae3cff71e3b41426045f659","revisions":{"c24e3bb4dfa85b51fae3cff71e3b41426045f659":{"kind":"REWORK","_number":1,"created":"2022-03-15 01:35:10.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/02/833702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/02/833702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/02/833702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/02/833702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/02/833702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/02/833702/1"}}},"commit":{"parents":[{"commit":"38c5e93a3fb13fbd3d51aca2f315d3f514477f4c","subject":"Prepare Yoga RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/38c5e93a3fb13fbd3d51aca2f315d3f514477f4c"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-03-15 01:06:49.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-03-15 01:12:59.000000000","tz":60},"subject":"Add $memcached_ipv6 in keystone/authtoken.pp","message":"Add $memcached_ipv6 in keystone/authtoken.pp\n\nWhen set to ``true`` (default is: ``false``) the\nkeystone::resource::authtoken will use the\ninet6_prefix_any function to set the ``inet6:``\nprefix for all emcached servers. This is useful\nfor services relying on python-memcached which\nrequire this format: ``inet6:[\u003chostname\u003e]:\u003cport\u003e``.\n\nRelated-Bug: #1964824\nDepends-On: I096be1bdb8429fb79693a681b709cae93cd8bbb8\nChange-Id: Iaa53db36697059121868f6415c7e55d0409c50f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c24e3bb4dfa85b51fae3cff71e3b41426045f659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c24e3bb4dfa85b51fae3cff71e3b41426045f659"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
