)]}'
{"id":"openstack%2Fpuppet-aodh~846276","triplet_id":"openstack%2Fpuppet-aodh~master~I643cc6b6cc85c1c3fc9d4803ebbf56ed1c40da5c","project":"openstack/puppet-aodh","branch":"master","topic":"param_typos","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-17 16:02:55.000000000","reason":"Takashi Kajinami replied on the change"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-06-20 19:12:28.000000000","reason":"Change was submitted"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2022-06-20 19:12:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I643cc6b6cc85c1c3fc9d4803ebbf56ed1c40da5c","subject":"Fix typos in parameter descriptions","status":"MERGED","created":"2022-06-17 07:32:24.000000000","updated":"2022-06-20 19:12:28.000000000","submitted":"2022-06-20 19:12:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846276-param_typos","meta_rev_id":"1db6552c1a3d2416cd7e8e08d99688184b49b849","_number":846276,"virtual_id_number":846276,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-20 19:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-06-17 16:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-06-20 17:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-06-20 17:52:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 08:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 14:17:04.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-06-17 14:17:04.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-06-20 17:52:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a6c4b0177eba7e03de794f584120fb826c95c8f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-17 07:32:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef4c5a299d8fecdfbd0b1f4bdc949bb0daca0ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 08:42:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8fd4ef7eabf49f68caa65f2992b3d12 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b66bd4f0f87b49c9bdebb59e5e58929c : SUCCESS in 4m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ea2a20653844d91b886bf6d09c825b0 : SUCCESS in 3m 20s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f54ce1cc7f743379c8296bcbaf1461c : SUCCESS in 12m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9e6ad60eeb0479bbb35ac88b7adfc86 : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da6302b025ec4d58b5fb9bf48c46b4a3 : SUCCESS in 10m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/664cf02067b942b69e330bf59d17dda3 : SUCCESS in 9m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f94c94b179342f48f48c024f8725e09 : FAILURE in 25m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/300825fc732e46329d54069b2490219c : SUCCESS in 55m 38s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8b7faa89d8948f489028edde1eb5ffe : FAILURE in 40m 16s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4232a6af3fc54c42ac5c74fbaa99525b : SUCCESS in 1h 07m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1977c0118f01464da65ada171802b7c7 : SUCCESS in 16m 55s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/921b9f55f94b47febc46036d7c4a05eb : SUCCESS in 22m 04s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ece1e393e3be4bc78f312a121ed043b2 : SUCCESS in 16m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64e53ba4afee4498bf61dd8f69a1bb49 : SUCCESS in 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18884fdba5366b04481eca77eea19d7a1744937c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-17 10:28:28.000000000","message":"Topic typos removed","accounts_in_message":[],"_revision_number":1},{"id":"ae88ab9d1cccb3a49a5c06b0d3f29cdd2aeaf8a4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-17 10:28:34.000000000","message":"Topic set to param_typos","accounts_in_message":[],"_revision_number":1},{"id":"fa846d192231320aedfc6abee5d5de8274555548","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-17 16:02:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e4d1212b3b9856c9b2219b68485fd20fc01dc1c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-20 17:52:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7141e5130adfec32c67860a6868fca00cd0da7a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:54:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fb6a38ef9411cdaee6e64b8395bccc10ec4bc01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 19:12:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e138be2d7144c25816b365a58fa7f67 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d1e1a832f8746559307680a344b6213 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/beac7a3e63c741f9a5c8c59816831597 : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/249703754b014317b0675e8048eeca0d : SUCCESS in 11m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faff8586d996421481ce721e00920245 : SUCCESS in 9m 10s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58fd0af98bda4e5db9b053c9369d4e2d : SUCCESS in 8m 32s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fd3c50ca6e5485d870ccf14139d20dc : SUCCESS in 1h 06m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d27fe0c94624d22ab570338cf6110ba : SUCCESS in 1h 13m 01s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ace8e5b2374474a87e46a0b7a1698f3 : SUCCESS in 23m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1db6552c1a3d2416cd7e8e08d99688184b49b849","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 19:12:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b29e4b5ab9131c5dd16a434a477e5edf2141531","revisions":{"8b29e4b5ab9131c5dd16a434a477e5edf2141531":{"kind":"REWORK","_number":1,"created":"2022-06-17 07:32:24.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/76/846276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/76/846276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/846276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/846276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/76/846276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/76/846276/1"}}},"commit":{"parents":[{"commit":"c58c36d92741bd41fda663d5765d9136399a77b5","subject":"Update section of the evaluation_interval parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c58c36d92741bd41fda663d5765d9136399a77b5"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-17 05:58:22.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-17 07:32:21.000000000","tz":330},"subject":"Fix typos in parameter descriptions","message":"Fix typos in parameter descriptions\n\nChange-Id: I643cc6b6cc85c1c3fc9d4803ebbf56ed1c40da5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
