)]}'
{"id":"openstack%2Fpuppet-aodh~849162","triplet_id":"openstack%2Fpuppet-aodh~master~Id133526308d84a45ba234b8dbef392aff784ade5","project":"openstack/puppet-aodh","branch":"master","topic":"config-acceptance","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-15 20:28:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id133526308d84a45ba234b8dbef392aff784ade5","subject":"Add acceptance tests for config management resources","status":"MERGED","created":"2022-07-09 03:16:40.000000000","updated":"2022-07-15 20:28:16.000000000","submitted":"2022-07-15 20:28:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849162-config-acceptance","meta_rev_id":"e3e7f1c36f25964ce7670ed69e27165ffb7a11ce","_number":849162,"virtual_id_number":849162,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 20:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-11 21:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-07-15 19:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-07-15 19:16:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 03:35:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 21:52:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-15 19:16:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"32db0038a5770b33df01e2c79c9f961c90acf7b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 03:16:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87b2c716ac4b0bea66edb20bc28df014b749d889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 03:35:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f4d227f36a34c71aa30dbc36def5557\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3344f478c56c408fb9c6a8dabd6b3733 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/267398382ce243fe9f3c268a0b05684b : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e45f2550542445e692fa7819d2f20af5 : SUCCESS in 2m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d8a541564654d54ae7b9cb076cce58f : FAILURE in 13m 25s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5184df83f96f46219112f2d6cd666d55 : FAILURE in 16m 48s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9af6c8baebbf4513a4336ca6a22b9178 : FAILURE in 12m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f592f1ba410474094b4a9579ab82afd : FAILURE in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebabedd47ed30ba973ded5025ac1eaada60157b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:00:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4b17c1004eb3b9fa9d7a3ad199661f17cab12bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:01:32.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b2fe67ff93a9e22b5820be1c430de8f63e4a7653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:02:25.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b6e3994295ba9509b60081aebc07e1216095a576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:06:58.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"0713fa6bcb77c1c0160cc5520c3889d12021640b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:19:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8f21e224c1f6dcdaaf5b42e3e41814fbed27065e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 04:45:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753fcc93ec6f49aa8a84400dd8f5c7d8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32c14d73685340e5b554186bc086cf6c : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ce3548e138141bfb0de8ca244566909 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c2364740dfe468f81098ae5e50b330a : SUCCESS in 2m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/422adf7ab0624d4e88a932cb84e3ccba : SUCCESS in 15m 41s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c1010fcb0904670844851aa930e44f5 : SUCCESS in 18m 14s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/122e8aacf57742c5b3584a7319f2caba : SUCCESS in 16m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd4b4e46bd14cf09ce0cd089a754cce : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d2c517d2948fb70cbeb1aca4e0971e45720309d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:03:49.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"75767e42fa04a7b94e6c1507b027020611b5a7d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:11:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5611c28e75446167be4a45eb6b8668b7408532a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 10:31:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/079590c384c84aa9a5f48253c4e83d8c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5b231e46af4a50a44b47e9c954912a : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b36f9791117e4f328ad8ea07ad16cb50 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/020d4ba76dd44c609e87bc79f64c84e0 : SUCCESS in 2m 58s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dad74894d52475abb8e2067d969c608 : SUCCESS in 14m 24s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b66deea0e184b86af8ca8b06c5a3fb6 : SUCCESS in 17m 24s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a40f85166cce412f9e868107628381e4 : SUCCESS in 13m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fca863ec77d401db837dfd31b21f874 : SUCCESS in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"69988b1ec4e227c61e675bbc79f770fa96a0f891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 12:28:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c984883552ae0228b808c5f5b8a562a1055f87fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 12:58:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c24277d438cc4fb0af9fe1368051d883\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dfa132c46394933a38286366e91e0f1 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/616070a2f6654324a973f09798f52c78 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e30ac73816f4b05a09aecb6c8e133f9 : SUCCESS in 2m 40s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be785bc4b80457b9316940b901b33f6 : SUCCESS in 15m 10s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/41bfb493dedd476c9612a13dd6d26987 : FAILURE in 28m 36s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/395ede80292649d7ad189cd71f3049db : SUCCESS in 11m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2ead609d5d84740b129d91283beb17c : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"534aac450d4bb82c109d33fab6e6d254f3b18826","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 13:00:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e746f4ce7d386056b9caa9235003fd2c864a9488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 13:17:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cff0e15432a4bc6929566e6d4081872\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55ca3c3c9e734b2f81a5059abf588ba6 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0c2e678e9dd41ec8f621ba9fb2f9712 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5123eb99a26246e5a8760c76ff889847 : SUCCESS in 3m 14s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fd7fc2eeb21448889f37a5a30cb4ab8 : SUCCESS in 15m 38s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/da09203802be48e188c18dbf95560c3a : SUCCESS in 13m 16s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56c7bdf326b440f293d84c2cd8a2e05d : SUCCESS in 10m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/056e7c5897d64a7abc41406c38410789 : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fde8276ebe25d1540d1815153e7d3091583cab8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-10 03:33:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"09e9caca6d520a7b0f58b038d8be371f5763b95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-10 03:57:48.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/641fd49f2edd4eeeaeeb5b5a5b01f8de\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eee78dbca1eb4dffa0cb1af8f8b6ef80 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7afdccdb657473290b296a159141184 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed71e08e5bba4a04bd3b56d0ca6aaf6a : SUCCESS in 3m 00s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/553edbe0271a4c56803a245941718bc5 : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9586c80b72884eac8b95afa8ae8cf7c6 : SUCCESS in 18m 06s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7da1cc49e2643039f4c605f37c63caf : SUCCESS in 13m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23a93644da574deb8a7dcbaf27c1903e : SUCCESS in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4dceda5e991d49351b0beb4a88f930fb8cd8f118","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 21:52:01.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"3a4eafff2c26d1efabbe140f34020d925f198604","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 19:16:25.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"d74dab7e80a84bf4174fe370dc36384d44d3c8ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 19:17:36.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1ccacdeb15eaa8f134ade954b3667f86129fb1da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:28:16.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1e99772519d4320a100b76c682f59c4\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d91738bcab541e6894245e94fd0d448 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3347166e5534006b221304401966326 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58771029250f431bbf5aa323c5cce346 : SUCCESS in 3m 43s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4946a652928b4cc991a8c2007b5f2204 : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":10},{"id":"e3e7f1c36f25964ce7670ed69e27165ffb7a11ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:28:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a1f73ef0402c0b869c0dcf4f9593527a7bceb75f","revisions":{"e1d4df89619df71e5c8d1f356340719c202b022d":{"kind":"REWORK","_number":1,"created":"2022-07-09 03:16:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/1"}}},"commit":{"parents":[{"commit":"8b29e4b5ab9131c5dd16a434a477e5edf2141531","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e1d4df89619df71e5c8d1f356340719c202b022d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e1d4df89619df71e5c8d1f356340719c202b022d"}]},"branch":"refs/heads/master"},"c9e874b304a363b62dbbee2afb7f8d0c7507e29b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-09 04:00:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/2"}}},"commit":{"parents":[{"commit":"328c55d839941a7bb1a8aec38676553dfad79208","subject":"api_paste_ini: Use ini_provider from openstacklib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/328c55d839941a7bb1a8aec38676553dfad79208"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:00:11.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c9e874b304a363b62dbbee2afb7f8d0c7507e29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c9e874b304a363b62dbbee2afb7f8d0c7507e29b"}]},"branch":"refs/heads/master"},"45e24a44ca81cff5404c4f3d4b9108f99020f711":{"kind":"NO_CHANGE","_number":3,"created":"2022-07-09 04:01:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/3"}}},"commit":{"parents":[{"commit":"e51001fd9a2dc7be3aa6bdc5b71d277e5adc9119","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e51001fd9a2dc7be3aa6bdc5b71d277e5adc9119"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:01:23.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/45e24a44ca81cff5404c4f3d4b9108f99020f711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/45e24a44ca81cff5404c4f3d4b9108f99020f711"}]},"branch":"refs/heads/master"},"1aaaa8c340b95ac44b30363a60922b4b8eef14bc":{"kind":"NO_CHANGE","_number":4,"created":"2022-07-09 04:02:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/4"}}},"commit":{"parents":[{"commit":"fa1a54515c3786acdd4f3527054dfef1e6c740b3","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fa1a54515c3786acdd4f3527054dfef1e6c740b3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:02:16.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1aaaa8c340b95ac44b30363a60922b4b8eef14bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1aaaa8c340b95ac44b30363a60922b4b8eef14bc"}]},"branch":"refs/heads/master"},"8999b6a2b64789d21c9d2a8e6ef5110493cbe94d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-09 04:06:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/5"}}},"commit":{"parents":[{"commit":"aba6134726065dfd8f97a31e51ce2f306083c38e","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/aba6134726065dfd8f97a31e51ce2f306083c38e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:06:48.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8999b6a2b64789d21c9d2a8e6ef5110493cbe94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8999b6a2b64789d21c9d2a8e6ef5110493cbe94d"}]},"branch":"refs/heads/master"},"65050894c41f757efbe13e34acc577f0feb97679":{"kind":"REWORK","_number":6,"created":"2022-07-09 04:19:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/6"}}},"commit":{"parents":[{"commit":"aba6134726065dfd8f97a31e51ce2f306083c38e","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/aba6134726065dfd8f97a31e51ce2f306083c38e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:19:35.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/65050894c41f757efbe13e34acc577f0feb97679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/65050894c41f757efbe13e34acc577f0feb97679"}]},"branch":"refs/heads/master"},"a6b78559beca57652ab7347debc48ccb764a2e85":{"kind":"NO_CHANGE","_number":7,"created":"2022-07-09 10:03:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/7"}}},"commit":{"parents":[{"commit":"31edf992554dad8e9567cbc5e0d496d98e091b1a","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/31edf992554dad8e9567cbc5e0d496d98e091b1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:03:49.000000000","tz":0},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a6b78559beca57652ab7347debc48ccb764a2e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a6b78559beca57652ab7347debc48ccb764a2e85"}]},"branch":"refs/heads/master","description":"Rebase"},"1823d12509e443b73bf7b87cc910fc4b4c71c221":{"kind":"REWORK","_number":8,"created":"2022-07-09 10:11:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/8"}}},"commit":{"parents":[{"commit":"31edf992554dad8e9567cbc5e0d496d98e091b1a","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/31edf992554dad8e9567cbc5e0d496d98e091b1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:11:41.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1823d12509e443b73bf7b87cc910fc4b4c71c221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1823d12509e443b73bf7b87cc910fc4b4c71c221"}]},"branch":"refs/heads/master"},"29b965bc149fce706fee9b2d21225d03759ec08c":{"kind":"REWORK","_number":9,"created":"2022-07-09 12:28:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/9"}}},"commit":{"parents":[{"commit":"31edf992554dad8e9567cbc5e0d496d98e091b1a","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/31edf992554dad8e9567cbc5e0d496d98e091b1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 12:28:05.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/29b965bc149fce706fee9b2d21225d03759ec08c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/29b965bc149fce706fee9b2d21225d03759ec08c"}]},"branch":"refs/heads/master"},"a1f73ef0402c0b869c0dcf4f9593527a7bceb75f":{"kind":"REWORK","_number":10,"created":"2022-07-10 03:33:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/849162/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/62/849162/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/62/849162/10"}}},"commit":{"parents":[{"commit":"31edf992554dad8e9567cbc5e0d496d98e091b1a","subject":"Switch provider to manage paste.ini file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/31edf992554dad8e9567cbc5e0d496d98e091b1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:16:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-10 03:33:01.000000000","tz":540},"subject":"Add acceptance tests for config management resources","message":"Add acceptance tests for config management resources\n\nChange-Id: Id133526308d84a45ba234b8dbef392aff784ade5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a1f73ef0402c0b869c0dcf4f9593527a7bceb75f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a1f73ef0402c0b869c0dcf4f9593527a7bceb75f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
