)]}'
{"id":"openstack%2Fpuppet-aodh~849170","triplet_id":"openstack%2Fpuppet-aodh~master~I56633cf5066f6608d1c61c19a956ebbad7cd9409","project":"openstack/puppet-aodh","branch":"master","topic":"bug/1981118","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-15 20:28:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56633cf5066f6608d1c61c19a956ebbad7cd9409","subject":"Switch provider to manage paste.ini file","status":"MERGED","created":"2022-07-09 04:00:20.000000000","updated":"2022-07-15 20:28:14.000000000","submitted":"2022-07-15 20:28:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849170-bug/1981118","meta_rev_id":"de9b5dafe9212923f68222591636789bf68c509a","_number":849170,"virtual_id_number":849170,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 20:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-11 21:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-07-15 19:16:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-07-15 19:16:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 05:17:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 21:51:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-15 19:16:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"83d4f291efd7bc82a139e5addeb642baa0a995ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:00:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b118426126ad856164f9c59f84f395acbd9be37","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:00:31.000000000","message":"Topic config-acceptance removed","accounts_in_message":[],"_revision_number":1},{"id":"005741369d4e966401791d43bcb6bd5b88aa2636","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:00:38.000000000","message":"Topic set to paste-ini-provider","accounts_in_message":[],"_revision_number":1},{"id":"3864ced5ff9b1787194af20864854508201a9c21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:01:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"80e9506d969f09e2eadc4391813899496605980c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:02:25.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"aee664094d45fd8ba1f9ddd8fe4c772995ac922c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:02:27.000000000","message":"Topic config-acceptance removed","accounts_in_message":[],"_revision_number":3},{"id":"dfcaf47123d1a581c9bcfbdf3732f5bdba2d8af1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:02:35.000000000","message":"Topic set to paste-ini-provider","accounts_in_message":[],"_revision_number":3},{"id":"aa1ff118c963819826837ab939a1f535fa25e69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:06:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c64a285a96435d80a11ce7fde629c2092f0d6cd2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:07:04.000000000","message":"Topic config-acceptance removed","accounts_in_message":[],"_revision_number":4},{"id":"cd7e53fe63fa35858b730cf2f0c3106f1809d1bb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:07:38.000000000","message":"Topic set to paste-ini-provider","accounts_in_message":[],"_revision_number":4},{"id":"9d65b8769bae5235941f36d8e2d01d895eddc2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 05:17:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb6c2fff4a44fdc974846ed90d4f569\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1abd479fbb36485f80e1602dc0b77582 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49bc500828774f6bb4e3ac0811a105e6 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddb7201f493a43ffbab6e6ca87e24505 : SUCCESS in 2m 56s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/19973812f763420bb1475506e2fb414b : SUCCESS in 10m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46d5980e6f584016921bab841196f569 : SUCCESS in 10m 17s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/714c34cf117d46ba852458148b2fd67d : SUCCESS in 7m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9bc8b73918645aaa6809a58c3c8d1c3 : SUCCESS in 7m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37d84b084b674255a36f30f0f2e059c7 : FAILURE in 24m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7551d17eaedf4bca9ce607593a4ed755 : SUCCESS in 53m 13s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dece356d11542e6b1d56ed3ed66425d : FAILURE in 40m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/156394a8d7674085a17efee52a68e66e : SUCCESS in 1h 09m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18136ea6a46f4e57b099432992a2e45e : SUCCESS in 12m 41s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e022f60e2dcc4ff98e28e7425455c10c : SUCCESS in 18m 42s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8625bd169904de292a9ae21a5670383 : SUCCESS in 14m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e31a5a76d6604f408746038f25f24f13 : SUCCESS in 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4569a947ca17040e9900b7251f37b2fdc0176f3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:03:18.000000000","message":"Topic paste-ini-provider removed","accounts_in_message":[],"_revision_number":4},{"id":"459dc35311c08eb59017f1c607b730070e6a25d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:03:21.000000000","message":"Topic set to bug/1981118","accounts_in_message":[],"_revision_number":4},{"id":"b289e6af8bf1e44fa172cee9902c5f65a18baa2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:03:37.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6a89bc3556241b6ff4324c1cff122047ac1b9abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 11:13:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0babc4d4f34739ab7c0ca4b1578efe\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68664e4039a74170bcc0beb536fd69e3 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b615cb2b3134a27b74611c3bf2678e3 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56eec1af74614c7b9958d5c9e761024d : SUCCESS in 4m 44s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ff0a24ddeb541cabbc3fad9f5e0f130 : SUCCESS in 9m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/234be98c5bcb426ea05642e7e8985eb4 : SUCCESS in 10m 28s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/add5bb822075423a90a307d5aebb386e : SUCCESS in 7m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aef770b84cfb45a291863048ca96959b : SUCCESS in 7m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcb1c73312dc4a86a6f5ff10a174fd23 : FAILURE in 24m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d9c9bf0888b4015b8fc9e0288dcb858 : SUCCESS in 57m 39s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a7af1f9673c428f9bef8a3ef9d271a8 : FAILURE in 32m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0f6208cd8f44180aa3346ea45fb5576 : SUCCESS in 1h 05m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f81fc7526ec47f18e7df73aeb67d992 : SUCCESS in 13m 17s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0a5d3447be44a01a5a1f0ec6c9b44c9 : SUCCESS in 17m 18s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9597bb54a24d42fb8e1efc7f0207f8e3 : SUCCESS in 14m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c183b184cea14a56aa370a80c1881036 : SUCCESS in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7574cab2a30fff10b0f36d79c2c9b4db777a921a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 21:51:56.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d39df6fa8f45ba833935fe9e3fa497a0989d5162","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 19:16:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b5a5d863d42dce641465162f638fe433a1531753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 19:17:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dd4dc512dcd9eba8707244966399189e59299f7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:28:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26764af253d74d7092228f57552f4efb\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/650e5cc1697c45889e6326161f889a47 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03c27204c78e462d9ba98e79196d0f36 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40c589ce1de24de2ba70a261b6980ab1 : SUCCESS in 3m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2157d9f2e0c0463dba9e154d2a6b8371 : SUCCESS in 17m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1049e1187de4c238a655aea4e937462 : SUCCESS in 18m 05s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf8a1f04d084f27bdf8db230c66c525 : SUCCESS in 10m 48s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a259226f13b24e6987f8bbcb598710a5 : SUCCESS in 58m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38a78fedcada42a3a24913e432e67165 : SUCCESS in 1h 06m 30s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/246f39ba00064c3f903c5a8bab1af9a0 : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":5},{"id":"de9b5dafe9212923f68222591636789bf68c509a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:28:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"31edf992554dad8e9567cbc5e0d496d98e091b1a","revisions":{"328c55d839941a7bb1a8aec38676553dfad79208":{"kind":"REWORK","_number":1,"created":"2022-07-09 04:00:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/849170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/70/849170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/1"}}},"commit":{"parents":[{"commit":"8b29e4b5ab9131c5dd16a434a477e5edf2141531","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:58:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:00:11.000000000","tz":540},"subject":"api_paste_ini: Use ini_provider from openstacklib","message":"api_paste_ini: Use ini_provider from openstacklib\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe aodh_config resource type.\n\nChange-Id: I56633cf5066f6608d1c61c19a956ebbad7cd9409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/328c55d839941a7bb1a8aec38676553dfad79208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/328c55d839941a7bb1a8aec38676553dfad79208"}]},"branch":"refs/heads/master"},"e51001fd9a2dc7be3aa6bdc5b71d277e5adc9119":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-09 04:01:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/849170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/70/849170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/2"}}},"commit":{"parents":[{"commit":"8b29e4b5ab9131c5dd16a434a477e5edf2141531","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:58:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:01:05.000000000","tz":540},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe aodh_config resource type.\n\nChange-Id: I56633cf5066f6608d1c61c19a956ebbad7cd9409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e51001fd9a2dc7be3aa6bdc5b71d277e5adc9119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/e51001fd9a2dc7be3aa6bdc5b71d277e5adc9119"}]},"branch":"refs/heads/master"},"fa1a54515c3786acdd4f3527054dfef1e6c740b3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-09 04:02:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/849170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/70/849170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/3"}}},"commit":{"parents":[{"commit":"8b29e4b5ab9131c5dd16a434a477e5edf2141531","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:58:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:01:52.000000000","tz":540},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe resource types to manage the \u003cservice\u003e.conf files.\n\nChange-Id: I56633cf5066f6608d1c61c19a956ebbad7cd9409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fa1a54515c3786acdd4f3527054dfef1e6c740b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/fa1a54515c3786acdd4f3527054dfef1e6c740b3"}]},"branch":"refs/heads/master"},"aba6134726065dfd8f97a31e51ce2f306083c38e":{"kind":"REWORK","_number":4,"created":"2022-07-09 04:06:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/849170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/70/849170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/4"}}},"commit":{"parents":[{"commit":"8b29e4b5ab9131c5dd16a434a477e5edf2141531","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:58:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:06:45.000000000","tz":540},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe resource types to manage the \u003cservice\u003e.conf files.\n\nChange-Id: I56633cf5066f6608d1c61c19a956ebbad7cd9409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/aba6134726065dfd8f97a31e51ce2f306083c38e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/aba6134726065dfd8f97a31e51ce2f306083c38e"}]},"branch":"refs/heads/master"},"31edf992554dad8e9567cbc5e0d496d98e091b1a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-07-09 10:03:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/849170/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/70/849170/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/70/849170/5"}}},"commit":{"parents":[{"commit":"8b29e4b5ab9131c5dd16a434a477e5edf2141531","subject":"Fix typos in parameter descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8b29e4b5ab9131c5dd16a434a477e5edf2141531"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 03:58:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:03:37.000000000","tz":0},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe resource types to manage the \u003cservice\u003e.conf files.\n\nCloses-Bug: #1981118\nChange-Id: I56633cf5066f6608d1c61c19a956ebbad7cd9409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/31edf992554dad8e9567cbc5e0d496d98e091b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/31edf992554dad8e9567cbc5e0d496d98e091b1a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
