)]}'
{"id":"openstack%2Fpuppet-aodh~850712","triplet_id":"openstack%2Fpuppet-aodh~master~Id71237bc97e04f6c83403f60605d3bff44bc944a","project":"openstack/puppet-aodh","branch":"master","topic":"cs9-pg","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-25 23:27:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Id71237bc97e04f6c83403f60605d3bff44bc944a","subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","status":"MERGED","created":"2022-07-22 04:37:26.000000000","updated":"2022-07-25 23:47:04.000000000","submitted":"2022-07-25 23:47:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850712-cs9-pg","meta_rev_id":"78fd36101ce077a2e1124eefa58d742862adb4ae","_number":850712,"virtual_id_number":850712,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-25 23:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-25 23:27:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-07-25 14:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-07-25 23:27:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 05:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 14:41:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"775b03807a3c92fca00af1846d4dcec3855eb94b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-22 04:37:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06b9e83a60e808009a5cb83af2948f9fb3f6d71a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-22 05:04:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e0e5466796e6db0f85530813a3b1ee8cd3796edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 05:04:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c40be5d55f3648d6bd87350ec454222d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f87c56a1e4643168ceb2b44b87befe9 : SUCCESS in 7m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78ed1db66ff04f999375b41250391519 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8f9e965e6fe427ba40c536aa1fbd3b5 : SUCCESS in 5m 31s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/558eef2ea69b4e86bcacc569e6822d89 : FAILURE in 24m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11dc5e6a45c54cbbb6c9f7aae8fe2d35 : FAILURE in 10m 52s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65930b9fd3ae4a28a2a782f11b5a0fae : FAILURE in 10m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14e2d7e0647b4210b024e3b53f9a0abe : FAILURE in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a460df5a9428299cda2f95dc16a5e52d733d455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 05:31:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5419ed4894474a0db7def129e007b7ea\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8290c4a43de142679a0faecf2d13705a : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/958fab245cef4e7bbf7cd4f5887a196c : SUCCESS in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4955af48edff47bd87a47f43aa7c15e3 : SUCCESS in 3m 48s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/542c114fb8694e0d9c70bfbbf04af722 : FAILURE in 25m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5111cd9ee2f945de93181917709e1e4a : FAILURE in 10m 19s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf9279e9e4264b62978ca092145430cb : FAILURE in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1a2077ece6b4a008efbe223ce7b4660 : FAILURE in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f5786f1b517cab1020513c4f3865eac5e248794","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-22 08:51:41.000000000","message":"Topic set to cs9-pg","accounts_in_message":[],"_revision_number":2},{"id":"32f5b303b8b5ec8a8617c51af8905496315790e8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-22 08:51:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"697fc82a84e9cdd781a493103b3cab5167145e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 09:18:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65473ac9d01f4239ab78569b9fbe46b2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fa4efcbc1b94185878aa58123114fc9 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11cffaf666574e25abc47992302214d9 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbb2ea900e5648daa156f937e330ddde : SUCCESS in 3m 55s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d3e569d010c40c2be34ef1e7c690dcc : FAILURE in 24m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5cee30da0004993b4a965b38a544ee5 : FAILURE in 12m 09s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cc38c1458e944339545471fef1541ff : FAILURE in 12m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb2674a7c2ff4515aede7aeb9e756663 : FAILURE in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5defe49ad2421ae34faecb511ec8587ccb29b1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-23 14:31:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bd5df291067bc1a896c93680582c8b6709e0761d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 14:47:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/536da98dfb004b0bac759fe59141e4e8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0810af495a50438d9f20b6d60d163dc3 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3396755c2e64d27a966ffc3cd9b01db : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f02de34c24e44b7daf3317e075cbfa4b : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/badbc33fa7b3491799ba9a936cd5261d : FAILURE in 13m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abf441f68ba445be84ea418f5d0f17d4 : FAILURE in 9m 45s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19a932f700c947f8976e35ff88c050a1 : FAILURE in 9m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5ab565016594cf1881a31731fa0231c : FAILURE in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5b8d048705969b149e588025dcd8bab6e63ecbfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-23 15:20:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"62774b974c5aec18252c417455c02b264082fc84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-23 15:24:23.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6b2189fed98cde062c41fad1fabe5e0ccf764c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 15:49:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d46a6349ba418f9c4d754d94a6aade\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2a2c8c9ba2047189f1cedc233b4adf8 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f762c56dfecd40caa3f0fb1175d49380 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a974733333e419699f755f45348775e : SUCCESS in 3m 17s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9ea388df6324352a2ed13e9f577be26 : SUCCESS in 17m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e91bb75bf004d05a0888391d3a9bc47 : SUCCESS in 10m 38s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4e64c8bfa5a459f850b0147673a1d0d : SUCCESS in 18m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21db443ff8a84e02b61776acab935b42 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f7542ba6755a0d319a9973597ae67b2d8b46af96","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-25 14:41:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b76aec659be184913572c3f8c975458b983c4bde","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-25 23:27:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"49891dcfd1e2287757aa7a096dbc8953bb9d1aca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:28:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7f69bc0728d96e66c109e6a15e9911347422d7bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:47:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37d6a85760f146a096399f357b92a87f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3363788669f4dc9983269eb9c84ae72 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e75be5a4554c4f438534b3bdd5a2038a : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf6ece7de6647a29117eaa567e6ccbc : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b785542e334462991a20e3344548d5e : SUCCESS in 13m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42741ad6a68547f39777348ebb49e030 : SUCCESS in 12m 12s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/447ccc6b7c8c48e0b0d5886b63361825 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":5},{"id":"78fd36101ce077a2e1124eefa58d742862adb4ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:47:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"358924dbca5ec0ff467c8b659893fd8e6e396211","revisions":{"049dad33773d0600a659835206938ddfe6325c14":{"kind":"REWORK","_number":1,"created":"2022-07-22 04:37:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/850712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/12/850712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/1"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-22 04:35:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-22 04:37:16.000000000","tz":540},"subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","message":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"\n\nThis reverts commit 0f1d26cdb67f4561be291db65cd4574ac0615cb5.\n\nReason for revert:\npuppet-postgresql 8.1.0 was released and now the module supports RHEL 9\n(and CentOS 9 effectively).\n\nDepends-on: https://review.opendev.org/850705\nChange-Id: Id71237bc97e04f6c83403f60605d3bff44bc944a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/049dad33773d0600a659835206938ddfe6325c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/049dad33773d0600a659835206938ddfe6325c14"}]},"branch":"refs/heads/master"},"ed6d4704654bd034842c7fb1270b87cf3431b720":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-22 05:04:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/850712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/12/850712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/2"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-22 04:35:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-22 05:04:24.000000000","tz":0},"subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","message":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"\n\nThis reverts commit 0f1d26cdb67f4561be291db65cd4574ac0615cb5.\n\nReason for revert:\npuppet-postgresql 8.1.0 was released and now the module supports RHEL 9\n(and CentOS 9 effectively).\n\nDepends-on: https://review.opendev.org/850705\nDepends-on: https://review.opendev.org/850713\nChange-Id: Id71237bc97e04f6c83403f60605d3bff44bc944a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ed6d4704654bd034842c7fb1270b87cf3431b720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ed6d4704654bd034842c7fb1270b87cf3431b720"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ac6e11656534a9cc06d7e0139dff4499611b551a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-23 14:31:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/850712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/12/850712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/3"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-22 04:35:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-23 14:31:13.000000000","tz":0},"subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","message":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"\n\nThis reverts commit 0f1d26cdb67f4561be291db65cd4574ac0615cb5.\n\nReason for revert:\npuppet-postgresql 8.1.0 was released and now the module supports RHEL 9\n(and CentOS 9 effectively).\n\nDepends-on: https://review.opendev.org/850705\nDepends-on: https://review.opendev.org/850790\nDepends-on: https://review.opendev.org/850713\nChange-Id: Id71237bc97e04f6c83403f60605d3bff44bc944a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ac6e11656534a9cc06d7e0139dff4499611b551a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ac6e11656534a9cc06d7e0139dff4499611b551a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5a82b8ebfd3ffb144fc08d9a6e05bc34b3469737":{"kind":"REWORK","_number":4,"created":"2022-07-23 15:20:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/850712/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/12/850712/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/4"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-22 04:35:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-23 15:20:11.000000000","tz":540},"subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","message":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"\n\nThis reverts commit 0f1d26cdb67f4561be291db65cd4574ac0615cb5.\n\nReason for revert:\npuppet-postgresql 8.1.0 was released and now the module supports RHEL 9\n(and CentOS 9 effectively).\n\nNote:\nThis change adds the service_provider fact because the latest\npuppet-postgresql requires the fact in CentOS/RHEL.\n\nDepends-on: https://review.opendev.org/850705\nChange-Id: Id71237bc97e04f6c83403f60605d3bff44bc944a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5a82b8ebfd3ffb144fc08d9a6e05bc34b3469737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5a82b8ebfd3ffb144fc08d9a6e05bc34b3469737"}]},"branch":"refs/heads/master"},"358924dbca5ec0ff467c8b659893fd8e6e396211":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-07-23 15:24:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/850712/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/12/850712/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/12/850712/5"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-22 04:35:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-23 15:23:49.000000000","tz":540},"subject":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"","message":"Revert \"CentOS 9: Disable unit tests dependent on puppet-postgresql\"\n\nThis reverts commit 0f1d26cdb67f4561be291db65cd4574ac0615cb5.\n\nReason for revert:\npuppet-postgresql 8.1.0 was released and now the module supports RHEL 9\n(and CentOS 9 effectively).\n\nNote:\nThis change adds the service_provider fact in test fact data because\nit is required by puppet-postgresql.\n\nDepends-on: https://review.opendev.org/850705\nChange-Id: Id71237bc97e04f6c83403f60605d3bff44bc944a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/358924dbca5ec0ff467c8b659893fd8e6e396211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/358924dbca5ec0ff467c8b659893fd8e6e396211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
