)]}'
{"id":"openstack%2Fpuppet-aodh~851651","triplet_id":"openstack%2Fpuppet-aodh~master~I9261c53fdb930fd9ff5e2c06e91eb312d665bb9a","project":"openstack/puppet-aodh","branch":"master","topic":"bug/1983300","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-08 17:00:10.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I9261c53fdb930fd9ff5e2c06e91eb312d665bb9a","subject":"Adapt to new type validation in puppetlabs-apache","status":"MERGED","created":"2022-07-30 13:02:40.000000000","updated":"2022-08-08 19:25:54.000000000","submitted":"2022-08-08 19:25:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851651-bug/1983300","meta_rev_id":"eb90d251eed5c2cf7238f9be1ba4e00d6d03ae54","_number":851651,"virtual_id_number":851651,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 19:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-08-02 12:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-08 15:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-08 15:17:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-30 14:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 12:27:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-08 15:17:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"244e0991c1aa46fa7598e41396cc5e570a28afee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 13:02:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d73f32c1e353e5d354ce5ad0ee9c6bfebcfafcda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 13:07:27.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"381fd0b37a963931c289af634b7e990aadfe9e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 13:08:54.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e81db5231bf23666b1186c5d5bdfd3f0d5ee9747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 13:14:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8dbecfaa0b8c51e600b49d7fbbd60a322727e87d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 13:41:27.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"15d1d47417f1ac5cd621f83c484eb617984a81d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 14:37:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6977e23b17ad4fef8b787ce90cb18558\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0c07aff57804afb9dcceabfa0b03f2e : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/221372eff7cb4a26a04ea3f78d2bf081 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b2cbd59c1de4e8a98bb9ab1d22b5c24 : SUCCESS in 3m 20s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab7287fe4c704c36a0f1509d20f94282 : FAILURE in 14m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aaa5c58b97a472089ae76b72c0575de : FAILURE in 10m 44s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f25b4389e6b54925be0ba031e4fcac3c : FAILURE in 9m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c72039eabfbc4716b4eb39c4d4dbe164 : FAILURE in 9m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/701ea182f2b448ec954f4e06727842d9 : FAILURE in 44m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71cd4d5e288145a78009aa0bf0204480 : SUCCESS in 1h 05m 04s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4180e011e4f247a1ba1498e0f39f8f7a : FAILURE in 31m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26824baef40f4e1ebfa1fa37bc0ef574 : SUCCESS in 1h 20m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/596f86f52e71454eaaa621eeb0d78034 : FAILURE in 20m 28s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee057b20564e4646bcd028ee85b316f7 : SUCCESS in 27m 13s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df5c9ffb7f44c45a1425f589733bbef : FAILURE in 20m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30ba0dcddfef4d5f8aac1de5fd056946 : FAILURE in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6487b4bf74348b30678d38957fd98470fbb6e855","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-02 05:39:02.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"fe1dad5ecccc7c5ffc077e51496331a91a6f11c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 07:03:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d8350f0d6d47d888086371ecfd00d8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2569886f3284fc68c4a3dba20808be5 : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51749361e4174e168dfecb7178bc5f1a : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90d5a8c8d499418eae837e0189166629 : SUCCESS in 4m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/78afa76fa2e5466bb588e56337225fdd : SUCCESS in 17m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f9b6b35e9dc47739a0646c8f9d8cb6c : SUCCESS in 11m 17s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e2e583bc891439fb28f973a32825970 : SUCCESS in 11m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53807767ea1e48b5ab8a958848c8bd48 : SUCCESS in 10m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07c246a3eeed4a53adf4eee6e3e71364 : FAILURE in 40m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/feac51af76984d9289d3b0f6ba68ac39 : SUCCESS in 58m 52s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8349bee00804924a615aeae8f89e12c : FAILURE in 44m 06s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/265c6a93ff0b4080a2712d46a5cdbef8 : SUCCESS in 1h 17m 57s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb7f47aba94843d181f3db4dfdbd6122 : FAILURE in 21m 36s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/46015fdd12414d189ad1e6a76d9eb721 : SUCCESS in 22m 31s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e86ba173db74e3d8ac1fdc2b391eb98 : FAILURE in 26m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2730bfea48ff4db083f02c0fe95ec71d : FAILURE in 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"53a1ecd9662d7b74eff33b46df3aad4848dcf484","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-02 12:27:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fc431832edc798d7452b99572bdf9def305e3e6a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-08 15:17:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3f2136c95366b3c484642a28f1de147011652662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 15:17:21.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba4415d46d54490ea3ef10ecfe704dab\n\n\nWarning:\n  Change 851652 in project openstack/puppet-openstacklib does not share a change queue with 851651 in project openstack/puppet-aodh","accounts_in_message":[],"_revision_number":5},{"id":"a78b970d940ca3eed5b850119ca83a6180fd68ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-08 17:00:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a0829ab206826ea173fa8a580919b29c834087f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 18:13:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63d4311301e547ac8eef413250057517\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2b036f140fa451cabb2abfebf8e22c9 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b18947bdad5942468bb1c1493036d281 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ebe2875cd06451bbc261c773d8b8e0a : SUCCESS in 3m 29s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ca16d0743d346d1af16f4bff5a79ff3 : SUCCESS in 17m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dfa19919ecc46fbb210003c435f7489 : SUCCESS in 10m 44s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b9de5baa8d74f4398b99fd87c63f5d0 : SUCCESS in 11m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f76b9f90b1f54d75a41188ea4354ac2e : SUCCESS in 10m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed9389afebf5451e83fd537a321eef28 : FAILURE in 36m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cee8c941743f4f819497d2214352a8ca : SUCCESS in 1h 04m 40s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6525e69afbe84acabb917a36c6436e3b : FAILURE in 43m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/705af0a7be9e443fad5474866ae236e8 : SUCCESS in 1h 10m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ac14e02b2f94566846214e90d0c7789 : FAILURE in 18m 26s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/059e59fb89994b0ea026d8749f923437 : SUCCESS in 27m 46s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78cae76cc69d4cd7baf20fd1e09960c4 : FAILURE in 18m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/195a9d90a32744c78f890693a4cb34a7 : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d024692d04dd965b889e24b41709f7481df44774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 18:14:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bddad50524f55b7bf015841e7083cc6363fe1ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 19:25:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e21745205104dacb4df73fc2330104f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/471dcc8295be45328c1d2788ebbff1ce : SUCCESS in 5m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7841a84ab37645b4a62c5ccc94582799 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b242d02c4d34460aa465a95ef3534dd7 : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/03bbbd6588184543bbdb4661a38fbbc4 : SUCCESS in 23m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6982084f13004980b1e12e699ea02a8f : SUCCESS in 11m 05s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6738da7ac81741ba8da17eafae2aeb2d : SUCCESS in 18m 42s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11a312c6916142f8b6471979d77a6648 : SUCCESS in 52m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/326f762e343e484e82ccb28a64694163 : SUCCESS in 1h 09m 58s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c775281215f346d6bb6f8e6c79032399 : SUCCESS in 21m 50s","accounts_in_message":[],"_revision_number":5},{"id":"eb90d251eed5c2cf7238f9be1ba4e00d6d03ae54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 19:25:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"976196bce37726419ff50113d65041ff0dff8f56","revisions":{"969049285b54e167d732c318f2a08f741a6244e5":{"kind":"REWORK","_number":1,"created":"2022-07-30 13:02:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/851651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/51/851651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/1"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 12:52:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 12:58:36.000000000","tz":540},"subject":"Adopt to new type validation in puppetlabs-apache","message":"Adopt to new type validation in puppetlabs-apache\n\nThe puppetlabs-apache module is enforcing more strict data type\nvalidation[1].\n\nThis change updates the default values to adopt to that change.\n\n[1] https://github.com/puppetlabs/puppetlabs-apache/commit/f41251e336fa3e7c31c19968ccee74121cf71e47\n\nChange-Id: I9261c53fdb930fd9ff5e2c06e91eb312d665bb9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/969049285b54e167d732c318f2a08f741a6244e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/969049285b54e167d732c318f2a08f741a6244e5"}]},"branch":"refs/heads/master"},"120e4e37a8b28be922d39fe1ea8f48953e1752be":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-30 13:07:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/851651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/51/851651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/2"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 12:52:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 13:07:14.000000000","tz":540},"subject":"Adopt to new type validation in puppetlabs-apache","message":"Adopt to new type validation in puppetlabs-apache\n\nThe puppetlabs-apache module is enforcing more strict data type\nvalidation[1].\n\nThis change updates the default values to adapt to that change.\n\n[1] https://github.com/puppetlabs/puppetlabs-apache/commit/f41251e336fa3e7c31c19968ccee74121cf71e47\n\nChange-Id: I9261c53fdb930fd9ff5e2c06e91eb312d665bb9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/120e4e37a8b28be922d39fe1ea8f48953e1752be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/120e4e37a8b28be922d39fe1ea8f48953e1752be"}]},"branch":"refs/heads/master"},"5b65226654fc022385e7fb0bb949291ac283a5f1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-30 13:08:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/851651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/51/851651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/3"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 12:52:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 13:08:44.000000000","tz":540},"subject":"Adapt to new type validation in puppetlabs-apache","message":"Adapt to new type validation in puppetlabs-apache\n\nThe puppetlabs-apache module is enforcing more strict data type\nvalidation[1].\n\nThis change updates the default values to adapt to that change.\n\n[1] https://github.com/puppetlabs/puppetlabs-apache/commit/f41251e336fa3e7c31c19968ccee74121cf71e47\n\nChange-Id: I9261c53fdb930fd9ff5e2c06e91eb312d665bb9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5b65226654fc022385e7fb0bb949291ac283a5f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5b65226654fc022385e7fb0bb949291ac283a5f1"}]},"branch":"refs/heads/master"},"07c617d2d5c33f8b8da3c48b5c33a518f70f7e4b":{"kind":"REWORK","_number":4,"created":"2022-07-30 13:14:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/851651/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/51/851651/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/4"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 12:52:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 13:14:33.000000000","tz":540},"subject":"Adapt to new type validation in puppetlabs-apache","message":"Adapt to new type validation in puppetlabs-apache\n\nThe puppetlabs-apache module is enforcing more strict data type\nvalidation[1].\n\nThis change updates the default values to adapt to that change.\n\n[1] https://github.com/puppetlabs/puppetlabs-apache/commit/f41251e336fa3e7c31c19968ccee74121cf71e47\n\nChange-Id: I9261c53fdb930fd9ff5e2c06e91eb312d665bb9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/07c617d2d5c33f8b8da3c48b5c33a518f70f7e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/07c617d2d5c33f8b8da3c48b5c33a518f70f7e4b"}]},"branch":"refs/heads/master"},"976196bce37726419ff50113d65041ff0dff8f56":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-08-02 05:39:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/851651/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/51/851651/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/51/851651/5"}}},"commit":{"parents":[{"commit":"8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30","subject":"Merge \"Add acceptance tests for config management resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8dd89d5aa2d6c3b8519eb2137911a3a3509a3d30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 12:52:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-02 05:38:30.000000000","tz":540},"subject":"Adapt to new type validation in puppetlabs-apache","message":"Adapt to new type validation in puppetlabs-apache\n\nThe puppetlabs-apache module is enforcing more strict data type\nvalidation[1].\n\nThis change updates the default values to adapt to that change.\n\n[1] https://github.com/puppetlabs/puppetlabs-apache/commit/f41251e336fa3e7c31c19968ccee74121cf71e47\n\nCloses-Bug: #1983300\nDepends-on: https://review.opendev.org/851652\nChange-Id: I9261c53fdb930fd9ff5e2c06e91eb312d665bb9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/976196bce37726419ff50113d65041ff0dff8f56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/976196bce37726419ff50113d65041ff0dff8f56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
