)]}'
{"id":"openstack%2Fpuppet-aodh~852188","triplet_id":"openstack%2Fpuppet-aodh~master~I0e11987c7d121b2fb639274ed60181c56ce4f6d1","project":"openstack/puppet-aodh","branch":"master","topic":"cleanup-coordination","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-12 08:10:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e11987c7d121b2fb639274ed60181c56ce4f6d1","subject":"Remove deprecated aodh::evaluator::coordination_url","status":"MERGED","created":"2022-08-04 16:12:09.000000000","updated":"2022-08-12 08:11:24.000000000","submitted":"2022-08-12 08:10:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852188-cleanup-coordination","meta_rev_id":"c9b3a384f3c9d42db72c22d7e3ac592e757932bc","_number":852188,"virtual_id_number":852188,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 08:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-08-08 15:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-12 06:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-12 06:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 17:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 15:17:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-12 06:42:38.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ca8492cba890e7c3eb8d19d2b2e0d90afabc0752","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:12:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45aae7aea9cf243dc8ddf71e24f612ca3a74c5fa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:12:39.000000000","message":"Topic set to cleanup-coordination","accounts_in_message":[],"_revision_number":1},{"id":"75fe5f328e972ea2e6ac0efdf22ffec2054950f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:16:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc09caf26bcc617cecc784a81ac85b86f17e9e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 17:23:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc96e81e1dd41c9bd3d504537da12df\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf688b907de34839a4176c1eba5a65a6 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/420c2af54611484d91317d12784c3f65 : SUCCESS in 5m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a10e309b48ab436780def6846081d58e : SUCCESS in 5m 10s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7d8e7513c594edfaac005f37d4c1af7 : SUCCESS in 23m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2e1f8b1095d408b9abfc1a813026d86 : SUCCESS in 18m 27s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f712d5cf1ffa4b2eb855547671e4627b : SUCCESS in 18m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6773a7948ac24ec19daa151f5cfd3bd1 : SUCCESS in 10m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9a1c8f0ac5945c5b96ae4bc7b70d908 : FAILURE in 34m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cbd129305514d7cb603e5030c858593 : SUCCESS in 57m 33s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fa29b21e9d348479df2d3df0db0c5c0 : FAILURE in 49m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9bfc5bd79b54610aaea7b623a8b7f3d : SUCCESS in 1h 04m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ad6f4bf8da14ce4ba2b0233e36da939 : FAILURE in 20m 02s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/376b2db4d24f4ba0bc0d2662223170c9 : SUCCESS in 23m 41s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec6d261ee91740b7823ed60af54da08a : FAILURE in 18m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4c26a2cf98b4ac4a4da28641902480b : FAILURE in 19m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6d0fd4b3c9e42219a05a63c49279c85 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c32824ddea9687c3b8a31c2d457d33f0da5289ef","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-08 15:17:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a372ac75acabf6a4f55724e2437594825c895882","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-12 06:42:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"62e55c3c765c0bae44dbdb8ed32d047ff6ae14b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:43:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"244bb5a90054c6f6aa368c2f09878838a10f955f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:10:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88ef19065309420a811caa3371ec7353\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f40807c34eb4c1db65de4dcf0956ee0 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8b57910933649d8a2e1aee4303421ab : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6640e2e441f4764aad6bad6747efdb0 : SUCCESS in 3m 50s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e311e5ec5c74d5990d231f1a83c7b65 : SUCCESS in 17m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f6111573a744794af0d96565f61b856 : SUCCESS in 10m 44s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/142aa1308aee42d1a31e7a027089f955 : SUCCESS in 11m 38s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7b973c62fed4bfbb5e926c3d9d5088e : SUCCESS in 57m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ecc3fd156a849d2bbb4d31049934ba3 : SUCCESS in 1h 05m 34s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d77911076984e5da3099294bc944afd : SUCCESS in 24m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/467eda42e3eb465da85a2ce1866e4aba : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c94b42e48923d6537cbb0ccf9abf2e70bee00d92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:10:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c9b3a384f3c9d42db72c22d7e3ac592e757932bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 08:11:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15796aac2c246899e4df48f0662bfe9\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18af660fb456488baedc2cfbfcdeb38f : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97682cfa99b013d149f232e9075acd9f3f45140a","revisions":{"b9f57d8ba9c9ba2a707a0edeb3f1b249a29effc2":{"kind":"REWORK","_number":1,"created":"2022-08-04 16:12:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/852188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/88/852188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/1"}}},"commit":{"parents":[{"commit":"6b33b3226bdceeb275951e0c357ea656a32a54d3","subject":"Merge \"Expose headers option of apache::vhost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6b33b3226bdceeb275951e0c357ea656a32a54d3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:11:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:11:58.000000000","tz":540},"subject":"Remove deprecated aodh::evaluator::coordination_url","message":"Remove deprecated aodh::evaluator::coordination_url\n\n... because it was deprecated during Xena cycle[1] in favor of the new\naodh::coordination class.\n\n[1] 16091c8dd6f97f55fb7875e7843f3de7017a99ce\n\nChange-Id: I0e11987c7d121b2fb639274ed60181c56ce4f6d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b9f57d8ba9c9ba2a707a0edeb3f1b249a29effc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b9f57d8ba9c9ba2a707a0edeb3f1b249a29effc2"}]},"branch":"refs/heads/master"},"97682cfa99b013d149f232e9075acd9f3f45140a":{"kind":"REWORK","_number":2,"created":"2022-08-04 16:16:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/852188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/88/852188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/88/852188/2"}}},"commit":{"parents":[{"commit":"6b33b3226bdceeb275951e0c357ea656a32a54d3","subject":"Merge \"Expose headers option of apache::vhost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6b33b3226bdceeb275951e0c357ea656a32a54d3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:11:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:15:54.000000000","tz":540},"subject":"Remove deprecated aodh::evaluator::coordination_url","message":"Remove deprecated aodh::evaluator::coordination_url\n\n... because it was deprecated during Xena cycle[1] in favor of the new\naodh::coordination class.\n\n[1] 16091c8dd6f97f55fb7875e7843f3de7017a99ce\n\nChange-Id: I0e11987c7d121b2fb639274ed60181c56ce4f6d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/97682cfa99b013d149f232e9075acd9f3f45140a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/97682cfa99b013d149f232e9075acd9f3f45140a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
