)]}'
{"id":"openstack%2Fpuppet-aodh~853383","triplet_id":"openstack%2Fpuppet-aodh~master~I3a2ec7199638e00cebefe68570b3c4c999d15621","project":"openstack/puppet-aodh","branch":"master","topic":"puppet-jammy","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-03 04:09:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3a2ec7199638e00cebefe68570b3c4c999d15621","subject":"Testing jobs with Ubuntu 22.04(jammy)","status":"ABANDONED","created":"2022-08-17 02:28:09.000000000","updated":"2023-02-03 04:09:54.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c97266a3b56142b09d62de1741fb4caa0d8608a8","_number":853383,"virtual_id_number":853383,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-21 04:12:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2022-08-17 02:29:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 03:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35d0ab08874e24274533911821700d1c9acca15a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 02:28:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ffc0c0fcfb02d3ca25e14a789c5ac79e43af01c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 02:29:03.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"1ba8bc6628e71ed407746b583a5d09d1d2442ed2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 03:49:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e25f5921b1437cb5d33b0f51b06949\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81c4f51c7ebf4a2fad16121f65408dca : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3749762c25d4444c9a69600c903c675c : SUCCESS in 4m 08s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/effa13b74f4d4f5ea9016f845ee24ea6 : SUCCESS in 17m 08s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9087cdfef25c472294f5c9daa6fc4cf8 : SUCCESS in 10m 43s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6610c79182304c08b650ab8bda700278 : SUCCESS in 11m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08c25dcc03f2492baa90e5304c917ae0 : FAILURE in 10m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a102a1f874e648d080c60fcc623ed201 : SUCCESS in 57m 58s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed3ec33c723941ef8a592d63b148a802 : FAILURE in 9m 28s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6533a68536543c1bd3406e59bcbcea5 : SUCCESS in 1h 12m 42s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0170e434e19846f59b535f4528f23608 : SUCCESS in 27m 29s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f4b878d54154dc9a2730ebd0229846c : FAILURE in 10m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53c9b69a63e949af9943a958b7664419 : FAILURE in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58e62d8e432eb937aae7327fe3a52815b9ff5399","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-17 02:39:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c053c4d92ccd6a77e9b74fe0e03e84095460f9c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 02:40:12.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/puppet-openstack-integration for 848536,2","accounts_in_message":[],"_revision_number":1},{"id":"7c77fb4aaacd78d9bc6c742174486c7d2cf34907","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-17 04:50:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"050ce090bfda44cba7ce99ee1657950844d06b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 05:51:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604c7e5a8fec40cc9da119970995178a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a47625c0be8a41c89941f596c13581e2 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc25228e31be40b1a182c5342a7c0d00 : SUCCESS in 2m 34s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/185d4ab3d416499f9543f7a16a2bb811 : SUCCESS in 8m 55s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37f6c66193e04de0b67b0525f0276b72 : SUCCESS in 6m 23s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7f446cccc1547e4be9ef97030a01060 : SUCCESS in 6m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3904516519ec455ab8f1093a6deb6e8b : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d8c779d7ce246df899e8a0fd9410a57 : SUCCESS in 48m 29s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/829563cf5bd142cfb097003831633c11 : FAILURE in 7m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56fbf20f273248a5b27ff6c6e4319cc3 : SUCCESS in 58m 29s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/adfabe0617ba426486627a584dec4f63 : SUCCESS in 18m 52s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b867380967df41a6875bbe06eda9fd4b : FAILURE in 7m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ad3304b6da647d3a79642f181f01a53 : FAILURE in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6663cfb9494590490cf2c9de2a6662594e8a72c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-21 03:09:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e239e47829aa29f6dc8f57ede025824b280ae52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 04:12:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f062383338a454dbfd8b30dda6aac98\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91d8c8ab54f046c18a9b8ccb2ef8e781 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04a8f94e594a49e8b2069215f91e18f9 : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d707b53d89184b6587a1195573f99338 : SUCCESS in 11m 41s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d01e24aee9684518bfa7173a2509ca6d : SUCCESS in 6m 53s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58055697e74f41abbaf684a02615e914 : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26cfd34a8ea144e08d60dbdf566ad2ff : FAILURE in 36m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bb19ce539da4df0b5192839cd8194e7 : SUCCESS in 51m 08s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bcf256cacbf48149c110ad25e3c3b23 : FAILURE in 49m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cf2d92c6679464d948bcf1679ea7d32 : FAILURE in 1h 01m 14s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/743d0ca6bba84b69bb092440b68557df : SUCCESS in 16m 42s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0433ded5a02f4876a20361f9d245dec3 : FAILURE in 5m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a0ed3aa63d6470893f8e5f8b8f5360b : FAILURE in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c97266a3b56142b09d62de1741fb4caa0d8608a8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-03 04:09:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c6e0b6ae1efa4ec68afb8ff5dfab68212c08efd","revisions":{"9c6e0b6ae1efa4ec68afb8ff5dfab68212c08efd":{"kind":"REWORK","_number":1,"created":"2022-08-17 02:28:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/853383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/83/853383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/83/853383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/83/853383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/83/853383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/83/853383/1"}}},"commit":{"parents":[{"commit":"c64f4251c54ce3c3dafd14d8e583f992434a040f","subject":"Merge \"Remove deprecated aodh::evaluator::coordination_url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c64f4251c54ce3c3dafd14d8e583f992434a040f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 02:27:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 02:27:41.000000000","tz":540},"subject":"Testing jobs with Ubuntu 22.04(jammy)","message":"Testing jobs with Ubuntu 22.04(jammy)\n\nDepends-on: https://review.opendev.org/848536\nChange-Id: I3a2ec7199638e00cebefe68570b3c4c999d15621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9c6e0b6ae1efa4ec68afb8ff5dfab68212c08efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9c6e0b6ae1efa4ec68afb8ff5dfab68212c08efd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
