)]}'
{"id":"openstack%2Fpuppet-aodh~863210","triplet_id":"openstack%2Fpuppet-aodh~stable%2Fxena~I082c2e49a8ae103c57c69bec5c3a7f3840d845f6","project":"openstack/puppet-aodh","branch":"stable/xena","topic":"lp/1966382-stable/xena","hashtags":[],"change_id":"I082c2e49a8ae103c57c69bec5c3a7f3840d845f6","subject":"setuptools: Disable auto discovery","status":"MERGED","created":"2022-11-02 08:56:20.000000000","updated":"2022-11-02 10:36:15.000000000","submitted":"2022-11-02 10:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863210-lp/1966382-stable/xena","cherry_pick_of_change":835545,"cherry_pick_of_patch_set":2,"meta_rev_id":"b3a20b7537fccd0727b24d6f628985ff0b31f66a","_number":863210,"virtual_id_number":863210,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-02 10:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-11-02 10:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-11-02 10:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 08:56:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"CC"},{"updated":"2022-11-02 09:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ec441b4701b18d586cfcbf258926946ad3f07f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 08:56:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2e256799d272df92b744894d60ee44eb526451ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 09:00:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da9a57be59054e8e887adcd4a4240144\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2632c366f6545d39f38725eaf1d15f5 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f7c5ab03c30f7816ac8a7e7cabdf3b1eb97857a8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 10:32:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7125595f791e8b671cb5d1cfcbd0b0433b805c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 10:33:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32962a30b8b6b7a0904ae937f677fa49e1d86e66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 10:36:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/255cde875e1c46cfaadb80e307ac9d69\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/095d783c59aa41fa81afb2d114cf61f7 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b3a20b7537fccd0727b24d6f628985ff0b31f66a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 10:36:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2521ec80a44b0750c7a261e59909f5d6ba82d936","revisions":{"2521ec80a44b0750c7a261e59909f5d6ba82d936":{"kind":"REWORK","_number":1,"created":"2022-11-02 08:56:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/863210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/10/863210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/863210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/863210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/10/863210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/10/863210/1"}}},"commit":{"parents":[{"commit":"6e86b30cc10a17c4707ad7ad757380f0ba3f14c9","subject":"Use openstack_spec_helper from zuul checkout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6e86b30cc10a17c4707ad7ad757380f0ba3f14c9"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 07:49:30.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 08:56:20.000000000","tz":0},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I082c2e49a8ae103c57c69bec5c3a7f3840d845f6\n(cherry picked from commit ed325e53b64dbb7b588814b58ce38b154fdc2987)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2521ec80a44b0750c7a261e59909f5d6ba82d936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2521ec80a44b0750c7a261e59909f5d6ba82d936"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
