)]}'
{"id":"openstack%2Fpuppet-aodh~871405","triplet_id":"openstack%2Fpuppet-aodh~master~I8c02a3651cf6a749e1039b0a3a9b92800fb4a79e","project":"openstack/puppet-aodh","branch":"master","topic":"policy_default_rule","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-24 08:54:48.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8c02a3651cf6a749e1039b0a3a9b92800fb4a79e","subject":"Expose policy_default_rule","status":"MERGED","created":"2023-01-23 01:48:32.000000000","updated":"2023-01-24 11:31:16.000000000","submitted":"2023-01-24 11:30:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871405-policy_default_rule","meta_rev_id":"4370ac3fa9ef27a2b32a5cb865f7067c53a8e24f","_number":871405,"virtual_id_number":871405,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 11:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-01-23 17:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-01-23 19:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-01-23 19:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 02:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 17:56:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-01-23 19:38:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a7ebe565f9892f5027309ec3dc6f3faddf89753f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 01:48:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9fd22456000f149f72dfe8fb6bc0a7eea21bc56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 01:49:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d71454f8eb1451d48a596b9b6654f7e6f6664ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 02:59:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a04aa61128e34c3fa5f56be8f6a57454\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cec0e67b01d4fc1aaff3204962cb206 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ad116c15edf45498cf2c721a4c2615b : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/92f0ad0e527b4dfdb857c573236a6bfe : SUCCESS in 11m 39s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1557a45a356e40f8a7166a4232cbf9d0 : SUCCESS in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fe620b6593d48a79a4cdd482a8cd00a : SUCCESS in 7m 02s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24680a05166f47adb77aeafffa5797ef : SUCCESS in 45m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c74a771665416589842751d4b79c60 : SUCCESS in 1h 02m 24s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7731bf4aad164e5dab78b7afe5d18377 : FAILURE in 1h 03m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5de77b88b6c444ba6f30a5f466d36cf : SUCCESS in 1h 02m 21s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/42ce1ad2cb6b45b7999bce8aa3716ce0 : SUCCESS in 19m 18s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8e13c3a7c17437daeb5568ef9456ebb : FAILURE in 5m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5114da4b415648e7962901b3ada7ff6d : FAILURE in 6m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31af418fa7934608bfbe7ca5095c0f7a : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4d57ba6a087296b5cdb60ce11c0610a1d743b43e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 05:27:53.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4a77e1f4b79779f5de0fbaa78dbb212b0ad18e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 06:54:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afc775e014c7476fa63e49f76cd48daa\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce7bcf4d39154cba929b4c556cad27da : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36de5bd208d24c17b90c02575c8792cc : SUCCESS in 2m 41s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/24697fee91ca44539fd2603320c3f5ea : SUCCESS in 13m 10s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9beb27a9ce94725a070619f5790e6ee : SUCCESS in 10m 42s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f627dc482ab541c1b0f4d4fa95e9400b : SUCCESS in 6m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48095f389a6a4f1cb15f1a546e7ae72b : SUCCESS in 56m 39s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36d927eeeee248ed910a39996c752e1e : SUCCESS in 1h 01m 21s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/543c2691c5184eb69ce418b8eb553a72 : FAILURE in 1h 23m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b72e9443005b4a21a8f9843fc9f66b20 : SUCCESS in 1h 07m 20s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/abf8180aecd8490cb4d06b6af2a80268 : SUCCESS in 19m 38s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f6aaeb89ac84e84989ba87704b60177 : FAILURE in 6m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4a2773ab6964055abd6f0bf4b3f17f9 : FAILURE in 7m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a553f895a7c465caf2ad1051c916265 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"73b519890026a41e75f3f15d975b618b0756b8f0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-23 17:56:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"accc77bbf39ef21a0b494ccc644dcd267c1e79a8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-23 19:38:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c29797bc3861d79fcf170729fab3e04b7dda13df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:40:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e101b0748e7ee60161d9775fe72561ffdf98c9da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 20:51:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d70fc88d72c04049b601c9b6e4a09820\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0d9cf65e116479f9444e7e812d510bd : FAILURE in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18583723083041e98e22c089865bcff8 : FAILURE in 2m 49s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae59a6ff2f8d460f9d0d38c09af4f86c : FAILURE in 8m 06s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ebf26f176e04c39a0ad74711d1cbe29 : FAILURE in 5m 34s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15e9d7ea2c4243449d3f886b0fb6742d : SUCCESS in 56m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aef9d33b3443458aad9b355b6c6b93ae : SUCCESS in 58m 41s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7cc828fce2040a4bb95acea66351604 : FAILURE in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ebd877bdc244300a4ee1c07d2bf90ee : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e360e717a347ae5c54d8cae7cdc2cb8d9f4e108a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-24 08:54:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"009994b6e51d0f14f23faa5a84e7c061c5d0eb07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:18:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b21ac98b1349c7bd5619c12b79c430\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dde42c59fc82453fb345ff87c6478d59 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfd5c1da6da94cedac95aa7ec0188cef : SUCCESS in 3m 01s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/72a12b6195f94baab80d96186d89bcc8 : SUCCESS in 10m 11s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd2f2b2cafcc45cd90c307e271408532 : SUCCESS in 8m 10s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13f5f2acba364a82b3edeefa98372675 : SUCCESS in 7m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b453ea75d873451e97a3bcafdb8ecfee : SUCCESS in 43m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/daf66000385e488792bf20a79323dc0d : SUCCESS in 1h 01m 34s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d936feecd11741569705c80959851c6a : FAILURE in 51m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2771a75d9b5146fbb3de6364ad44c404 : SUCCESS in 1h 06m 12s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e59fe5a5ee3942b59565a6aef9623aa4 : SUCCESS in 22m 50s\n- puppet-openstack-litmus-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9a4d2ea5c4b40fc9c84c8098e24f3ae : SUCCESS in 13m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c71a50de6934193a65ba9d575f0ab56 : SUCCESS in 14m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a86bc434dadb418195af1753f00a93e3 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d07a2df91154c69981c5bd3d16ff8422ca1be405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:19:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9297f85190b3bc7fdbf025b686dc8dc78492ded2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:30:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f52e809265e478f93d2137ee3c7c238\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4eeb70e4df9041a79cbec47ece9ea76d : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb0d18d284a44f52816a0e580e12b5cd : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/506c7c694ce243379e42977e431faf35 : SUCCESS in 11m 39s\n- puppet-openstack-unit-7.16-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3426a5def6e24cf98d4a21a7d9625c5f : SUCCESS in 7m 33s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddf05909d97d4d32b2bfedaa1c63c79e : SUCCESS in 58m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3a48ad2fa374a47bb6d1eb1efbf44ea : SUCCESS in 1h 06m 24s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/43205d5840d84b06a4f096fb86e5e309 : SUCCESS in 18m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc0b0393b7f946e0b52cba89184dcbee : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"13d39ce625d017b46f5c20cd8a702dde6a2b651d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:30:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4370ac3fa9ef27a2b32a5cb865f7067c53a8e24f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:31:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64ee6ccb72344043845407afe83b15bd\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3402c4a25154a5f94b3fac07a6c4acb : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","revisions":{"3b4a90e7fa1e014e40783eb0a2e7eab39febf661":{"kind":"REWORK","_number":1,"created":"2023-01-23 01:48:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/871405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/05/871405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/1"}}},"commit":{"parents":[{"commit":"c2f586d2adc143c7cb0a6b586aeb65be8ef401c2","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c2f586d2adc143c7cb0a6b586aeb65be8ef401c2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 01:45:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 01:45:57.000000000","tz":540},"subject":"Expose policy_default_rule","message":"Expose policy_default_rule\n\nThe option has been managed by the underlying puppet-oslo module but\nhas not bee configurable. This introduces the parameter to customize\nthe option.\n\nChange-Id: I8c02a3651cf6a749e1039b0a3a9b92800fb4a79e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3b4a90e7fa1e014e40783eb0a2e7eab39febf661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3b4a90e7fa1e014e40783eb0a2e7eab39febf661"}]},"branch":"refs/heads/master"},"0834aa4e0cbd5aa93306ecdb1335850c3b905f44":{"kind":"REWORK","_number":2,"created":"2023-01-23 01:49:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/871405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/05/871405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/2"}}},"commit":{"parents":[{"commit":"c2f586d2adc143c7cb0a6b586aeb65be8ef401c2","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c2f586d2adc143c7cb0a6b586aeb65be8ef401c2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 01:45:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 01:49:47.000000000","tz":540},"subject":"Expose policy_default_rule","message":"Expose policy_default_rule\n\nThe option has been managed by the underlying puppet-oslo module but\nhas not bee configurable. This introduces the parameter to customize\nthe option.\n\nChange-Id: I8c02a3651cf6a749e1039b0a3a9b92800fb4a79e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0834aa4e0cbd5aa93306ecdb1335850c3b905f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0834aa4e0cbd5aa93306ecdb1335850c3b905f44"}]},"branch":"refs/heads/master"},"df3fce9e6b285d846ab4945cb24de7b51a3d425e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-23 05:27:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/871405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/05/871405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/05/871405/3"}}},"commit":{"parents":[{"commit":"c2f586d2adc143c7cb0a6b586aeb65be8ef401c2","subject":"Switch to Ubuntu Jammy (22.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c2f586d2adc143c7cb0a6b586aeb65be8ef401c2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 01:45:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-23 05:27:43.000000000","tz":540},"subject":"Expose policy_default_rule","message":"Expose policy_default_rule\n\nThe option has been managed by the underlying puppet-oslo module but\nhas not been configurable. This introduces the parameter to customize\nthe option.\n\nChange-Id: I8c02a3651cf6a749e1039b0a3a9b92800fb4a79e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
