)]}'
{"id":"openstack%2Fpuppet-aodh~875579","triplet_id":"openstack%2Fpuppet-aodh~master~I73444acddba8818dfcee5d2b2bca52c7e4cf3451","project":"openstack/puppet-aodh","branch":"master","topic":"validate-boolean","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-07 05:46:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I73444acddba8818dfcee5d2b2bca52c7e4cf3451","subject":"Add strict validation about boolean parameters","status":"MERGED","created":"2023-02-28 01:55:41.000000000","updated":"2023-03-07 08:18:30.000000000","submitted":"2023-03-07 08:18:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875579-validate-boolean","meta_rev_id":"27dde497dcdc97bf2d0b0a811370a848d733d2b1","_number":875579,"virtual_id_number":875579,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 08:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-03-03 08:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-03-06 15:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-03-06 15:20:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 03:14:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 05:55:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-03-01 17:48:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"9d1933b1ee983198151c7ac8d805414a986dc8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-28 01:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4b0ea6e57cf7300a8149fc9c911095f89906ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 03:14:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192d62e6a0f74d2cb1c0f293d637026c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc0a2da9aab4dbb9e3a49d6c8bb4dd5 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07e6db11ade14a6cab9c07558a77420a : SUCCESS in 2m 41s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed82b7ef85a74d0eb43525a7b973d9b5 : SUCCESS in 10m 15s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee04fc1a4f544e6588216f60b036185e : SUCCESS in 7m 32s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9a1bd599edc422cb78ed423faf7d9cb : SUCCESS in 7m 26s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/421790e27ab445309d31f0ac6b44af4e : SUCCESS in 47m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e147898ed4004432b1f30e201675c331 : SUCCESS in 1h 10m 38s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3344045c83d1400498d6c10897412ba8 : FAILURE in 57m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9806824b98b47ecae733b708e75c230 : SUCCESS in 1h 13m 50s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff2403c7888a4d579f724727d4168554 : SUCCESS in 18m 31s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6db2efd16e944fab00a9c43a287d2d9 : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af9ab24474f045359273ae23389c5a5e : SUCCESS in 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb183c9846ba0b9f4e201f67da004a0ab6755aa7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-28 04:05:35.000000000","message":"Topic boolean removed","accounts_in_message":[],"_revision_number":1},{"id":"8c592b777a319406b4ef0ed05a932f6566ee0185","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-28 04:05:40.000000000","message":"Topic set to validate-boolean","accounts_in_message":[],"_revision_number":1},{"id":"f59393f10f140cfe559de42c637139d27c47808d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-28 05:55:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f96bafa59b19f696680b18beac7d1ccfa7b89dd0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-01 17:48:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20f7aaeb0a584ca344612c891c0bf7bcad2191bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 17:49:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d396dbe2889f5aa2665671e2b418c67fe8afe65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:03:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6a0471acfb74d6183f08ab1d72c1da7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6a9a508fd9c405ba23f18b02a823ba5 : FAILURE in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc17db7b7314e6ab171315d6b8077f4 : SUCCESS in 2m 29s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ece662a950949dfb547926fc9e235c2 : SUCCESS in 11m 19s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3a7330126844f35a88c1325765aabc8 : SUCCESS in 7m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cec64ff3fd9e459d8f7993829516aff2 : SUCCESS in 58m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7648a578d991435db36eabee1984a61e : SUCCESS in 1h 11m 19s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4ab84e2ae584a0397f4497a5fb306b6 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5b43974c2e6b74aafbdceb9aa19e570d2ffda49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 04:19:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b05cf5cb7af9c77c599aed75c1d88b37ac37ed6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 04:29:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f20fb6468a451801a5acd56b23336a0bb57cb8cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-03 04:37:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4cc7246965407ee8f12d34430c42bb06d0412e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 05:50:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcfab3a5ef6b41238e08c2e0be4303a5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/067dbb6c3055469d8aa2431afcea6f46 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/519c2c7732df48d5bce99f514c80a68e : SUCCESS in 4m 59s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7af0ca5f497b4a54ad28b1a21b6de92d : SUCCESS in 13m 07s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76539d28669b44eaa9dcd16080a37160 : SUCCESS in 7m 03s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47db9830150f4348aa2c33127ef180ca : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86dfe099baec4c49807acf1adcf1c324 : SUCCESS in 44m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac938189eb3a406d8d500ac033ab8162 : SUCCESS in 1h 10m 25s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e9dd94d921a4038bfab6012d52bc5ea : FAILURE in 52m 56s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4a3a9ee3f0a4018b5d7470568233244 : SUCCESS in 1h 05m 48s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0defe06010064b02be9b283312ef6e9f : SUCCESS in 17m 49s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f528bc6db3414bc7b891de6e48d5cb37 : SUCCESS in 13m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcb0ceb8ec1b4925b5d94d29c86de48f : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c8513d29759f0b981f53598b4ca7b1d4efcf30a7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-03 08:15:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2792cd7e8c69a34cc99be8e722ae7d13fb69f32f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-06 15:20:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ca871796f1d65cee230a8b1879a8b3073dcf1bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 15:21:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5fa747256f38b59a4670e28175d8aa105c32f469","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 16:44:05.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61a15a91a1e8455ca68e5d9bbfa7eed5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/deb2142b7a9d4c168c10e1eda1f292a4 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d78c4e013477444da8176ce25f7ccc93 : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e029db583b34c3c9d9168cca90700e5 : SUCCESS in 10m 54s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/539f7d289f504c6eadda2d8f6fb282d8 : SUCCESS in 7m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/98fe985c39ab4a45a21d091726c9798f : FAILURE in 1h 00m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/76dcb4666e6c4bdfa2a5cec290b56fec : SUCCESS in 1h 18m 01s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/debc670219d04272819ceb838563f77a : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":4},{"id":"c4b08e723400eaee55448416326789f52062d357","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-07 05:46:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e13582d61f27b4d38b3b5bf16cdf75ccce11ae76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 06:59:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e3fd9eb25b453abd66d0c7e3381f46\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2e3149857634e499090229c11557d05 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5311416dca2944a39cff821578886ce5 : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5389e4abd6a2436aa99dab81467c321a : SUCCESS in 11m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad13c9c921684b098d4b6dec3a359c28 : SUCCESS in 7m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce0888b5245645688b639a4cd46486a4 : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cfdade67ccb4f7a91766c82465d96ec : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc6b0d708f6d4e4ebbbc8835d050cd7d : SUCCESS in 1h 01m 40s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f66da3f0481455ead9798c120919729 : FAILURE in 56m 47s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b575bf9b053947aa97f50cd5dd91eccd : SUCCESS in 1h 08m 22s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c347d0a404e942589d3bfe8e429ad21f : SUCCESS in 19m 32s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f66f3a3d0f64cb0aaa9e9fb64dab4b2 : SUCCESS in 12m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/805ea3b615c547839d5ff2847e76fd9a : SUCCESS in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a0b6c3811d3a3fd062eae680bed565aaf40c7d7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 07:00:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7bc1d57c7f15a52c8e3a693a6848dd224788e339","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 08:18:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eddafc7774fd493e8624b6d5821e4b63\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c73a2fcce9ab4f149c9038f485d15af7 : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/720a6f0f8a5d45aea05479385f489856 : SUCCESS in 2m 45s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/975a7ec9a3fa4edf8d91c942efb50e0f : SUCCESS in 11m 53s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7d52fd4174d4eb5928a552a28123896 : SUCCESS in 7m 42s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0aa5cb48c8574ba3ad35037f00e05b57 : SUCCESS in 57m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/817ead72f150476c80865d9c01e92502 : SUCCESS in 1h 13m 36s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb1c2668cb08499c9614127d60baaa1c : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":4},{"id":"27dde497dcdc97bf2d0b0a811370a848d733d2b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 08:18:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bea839fe5a88a0f37fc4b16d5e55943635e89d8b","revisions":{"19c163fa7013344f5c415c2e19ac2f1ab6b34041":{"kind":"REWORK","_number":1,"created":"2023-02-28 01:55:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/875579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/79/875579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/1"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 01:53:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 01:54:41.000000000","tz":540},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I73444acddba8818dfcee5d2b2bca52c7e4cf3451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/19c163fa7013344f5c415c2e19ac2f1ab6b34041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/19c163fa7013344f5c415c2e19ac2f1ab6b34041"}]},"branch":"refs/heads/master"},"6bbc4f5cfac171284acf778d1a90913c17a91606":{"kind":"REWORK","_number":2,"created":"2023-03-03 04:19:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/875579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/79/875579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/2"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 01:53:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-03 04:19:26.000000000","tz":540},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I73444acddba8818dfcee5d2b2bca52c7e4cf3451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6bbc4f5cfac171284acf778d1a90913c17a91606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6bbc4f5cfac171284acf778d1a90913c17a91606"}]},"branch":"refs/heads/master"},"898728661a592842463996067d08df49f916b236":{"kind":"REWORK","_number":3,"created":"2023-03-03 04:29:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/875579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/79/875579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/3"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 01:53:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-03 04:29:03.000000000","tz":540},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I73444acddba8818dfcee5d2b2bca52c7e4cf3451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/898728661a592842463996067d08df49f916b236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/898728661a592842463996067d08df49f916b236"}]},"branch":"refs/heads/master"},"bea839fe5a88a0f37fc4b16d5e55943635e89d8b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-03 04:37:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/875579/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/79/875579/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/79/875579/4"}}},"commit":{"parents":[{"commit":"3bf38f8403d95103f0af0928a679d8e47821a110","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3bf38f8403d95103f0af0928a679d8e47821a110"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-28 01:53:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-03 04:37:10.000000000","tz":0},"subject":"Add strict validation about boolean parameters","message":"Add strict validation about boolean parameters\n\nThis ensures the parameters used by if-else logic accept only boolean\nvalues because non-boolean can result in unexpected behavior.\n\nChange-Id: I73444acddba8818dfcee5d2b2bca52c7e4cf3451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bea839fe5a88a0f37fc4b16d5e55943635e89d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/bea839fe5a88a0f37fc4b16d5e55943635e89d8b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
