)]}'
{"id":"openstack%2Fpuppet-aodh~875796","triplet_id":"openstack%2Fpuppet-aodh~master~I2a52329948becfcfb799f034847db9eae7bd903d","project":"openstack/puppet-aodh","branch":"master","topic":"fact-lint","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-03 10:14:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a52329948becfcfb799f034847db9eae7bd903d","subject":"Replace legacy facts and use fact hash","status":"MERGED","created":"2023-03-01 03:03:18.000000000","updated":"2023-03-03 10:14:04.000000000","submitted":"2023-03-03 10:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875796-fact-lint","meta_rev_id":"1395d335ad0906da36ac171f185053e66d21abbb","_number":875796,"virtual_id_number":875796,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 10:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-03-01 17:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-03-03 08:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-03-03 08:15:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 08:54:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 17:48:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-03-03 08:15:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"64b5b7ac99ded8709de1bf8e37f9aaf747e467a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 03:03:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16e0d0390fffc3de4d9ba1ba88ce59e88fef0a6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:09:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07707b49e7e766d7bccf0a5c3af0755e5fbb5f23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:31:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f7e66239241ff466754c58abb34be7972384f12c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:31:47.000000000","message":"Topic boolean removed","accounts_in_message":[],"_revision_number":3},{"id":"cec71d6560fbdb0ebe90da64489d2ccf6714ca46","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:31:51.000000000","message":"Topic set to fact-lint","accounts_in_message":[],"_revision_number":3},{"id":"4992a4ed9ea1758f38d01e2bbc817cb6bc3cce08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:37:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c6566d81558012ac8949c147f8025f8368c24fc5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:37:57.000000000","message":"Topic boolean removed","accounts_in_message":[],"_revision_number":4},{"id":"57bd3598611919598c3162ef5f4816beb9edc842","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:38:02.000000000","message":"Topic set to fact-lint","accounts_in_message":[],"_revision_number":4},{"id":"c73c559643e6828e1f1ebbb1b1d377ed4de9fafb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 04:49:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"28e76d03f7052eb7133941bd3e1aa757c98204df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 05:44:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2eb3833ac484078f4fc93493c08f100670b14557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 06:06:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"514ceb36ea25a37cad429e91a2a2c15de5be70ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 06:59:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1ec200c25ef367ef22000cf82f3cee2ccae13b3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 07:00:57.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9b5f34317a0fbe44016747e2fe5c8b2d4bd31167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 07:14:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fbc8d8d4d5afe13b2ac5735c41e49c4c812829aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 07:17:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"111b3289ffd4b066ee402b9e4c913092abe1034d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-01 07:47:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1ecfa9003489d204e4cd6b537abc905fc4cd46bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 08:54:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf07d2add90940b9a38b9afe13c4fe5c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fb1c5cc28db4aa0adaa26749424d84d : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63af8096e449463ea45827918052f3c1 : SUCCESS in 3m 16s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6f1c1471a62452482ab8df62b48bae6 : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecc299550b634701ae4a5e911ec6dec8 : SUCCESS in 7m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d458b56256f47c59e7eead6c4951f46 : SUCCESS in 6m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/984f0dc036d34c08b1ff6d2b799a64e3 : SUCCESS in 49m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec13148a47cc41309c0d24d13f3688a1 : SUCCESS in 57m 51s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbe42bcb14754526bc3a563b6fc61a3b : FAILURE in 52m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fce4e929e9042d395df9c5686285db3 : SUCCESS in 1h 04m 28s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3b5349e084a46c19ffd7d3cafbd8b46 : SUCCESS in 18m 21s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02c437537622451684204c5c59b90dda : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c703ff6d7714a2db2ccf7ae460bf15f : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e9a38ce6a3dcfd5143ab78c76a6dcc675972ce50","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-01 17:48:48.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"1c9b0fd1bfe00acdae451dceb50548f4835916d2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-03 08:15:07.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"c83e8d58aabdd14f48ef00729a9cf9bbb4913a0d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-03 08:15:11.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"47a523b9600904dd5a8226c1807b6e160b98048b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 08:16:03.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"331f2dc0004d2601da7cf3ec12d8e3dc1e107c67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 10:14:04.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ae6cf8c20e844228f9a6577676bb5e7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9d18f7c85de4e3fa045d800a5548eb4 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a971b8d92b94746842a7da9fd1c8c33 : SUCCESS in 3m 20s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/99dda240b4324e08a344b0c742c0c39b : SUCCESS in 12m 53s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2871cd4adb644afab1a4eccbe0c1b17 : SUCCESS in 9m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f16223bca3b64f1899e7a341687be2d0 : SUCCESS in 1h 02m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f21e5b329ea43298149c473835b6191 : SUCCESS in 1h 51m 53s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7769786f1a824d1787e950431e4a9ae8 : SUCCESS in 23m 22s","accounts_in_message":[],"_revision_number":12},{"id":"1395d335ad0906da36ac171f185053e66d21abbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 10:14:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"3bf38f8403d95103f0af0928a679d8e47821a110","revisions":{"b1f8002bf9457b13a2348959c1293ba366894d2a":{"kind":"REWORK","_number":1,"created":"2023-03-01 03:03:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/1"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:03:05.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b1f8002bf9457b13a2348959c1293ba366894d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b1f8002bf9457b13a2348959c1293ba366894d2a"}]},"branch":"refs/heads/master"},"0551c1efb99512113cc6e2a038cbf50688819ee7":{"kind":"REWORK","_number":2,"created":"2023-03-01 04:09:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/2"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 04:08:52.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nDepneds-on: https://review.opendev.org/875799\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0551c1efb99512113cc6e2a038cbf50688819ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0551c1efb99512113cc6e2a038cbf50688819ee7"}]},"branch":"refs/heads/master"},"4bb491a8cd4a2755bf96dd9a6b6f33469a638e40":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-03-01 04:31:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/3"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 04:31:19.000000000","tz":0},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nDepends-on: https://review.opendev.org/875799\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4bb491a8cd4a2755bf96dd9a6b6f33469a638e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4bb491a8cd4a2755bf96dd9a6b6f33469a638e40"}]},"branch":"refs/heads/master","description":"Edit commit message"},"447e2bf717354c11a6b43d910946413ed73f86a8":{"kind":"REWORK","_number":4,"created":"2023-03-01 04:37:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/4"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 04:37:33.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nDepneds-on: https://review.opendev.org/875799\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/447e2bf717354c11a6b43d910946413ed73f86a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/447e2bf717354c11a6b43d910946413ed73f86a8"}]},"branch":"refs/heads/master"},"28bef5caf30831f43589ceda3e98a0d9d1c7ab26":{"kind":"REWORK","_number":5,"created":"2023-03-01 04:49:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/5"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 04:49:31.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nDepneds-on: https://review.opendev.org/875799\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/28bef5caf30831f43589ceda3e98a0d9d1c7ab26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/28bef5caf30831f43589ceda3e98a0d9d1c7ab26"}]},"branch":"refs/heads/master"},"8513ecf8f37caea5885751ad12ba52a620c9b8b7":{"kind":"REWORK","_number":6,"created":"2023-03-01 05:44:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/6"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 05:31:42.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8513ecf8f37caea5885751ad12ba52a620c9b8b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/8513ecf8f37caea5885751ad12ba52a620c9b8b7"}]},"branch":"refs/heads/master"},"15727a8139aa067610aee0bf482d523d88ce8f35":{"kind":"REWORK","_number":7,"created":"2023-03-01 06:06:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/7"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 06:06:48.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/15727a8139aa067610aee0bf482d523d88ce8f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/15727a8139aa067610aee0bf482d523d88ce8f35"}]},"branch":"refs/heads/master"},"44c9b53562625a481682f18f7d76a8fc7b10af68":{"kind":"REWORK","_number":8,"created":"2023-03-01 06:59:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/8"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 06:59:23.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/44c9b53562625a481682f18f7d76a8fc7b10af68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/44c9b53562625a481682f18f7d76a8fc7b10af68"}]},"branch":"refs/heads/master"},"7372cd1d920aafb05e5dcc2b2033d4cc6562d216":{"kind":"REWORK","_number":9,"created":"2023-03-01 07:00:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/9"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 07:00:49.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7372cd1d920aafb05e5dcc2b2033d4cc6562d216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7372cd1d920aafb05e5dcc2b2033d4cc6562d216"}]},"branch":"refs/heads/master"},"ea19a43e23e0256d84b4ef0c82b0cfb5bcee0fda":{"kind":"REWORK","_number":10,"created":"2023-03-01 07:14:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/10"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 07:14:46.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ea19a43e23e0256d84b4ef0c82b0cfb5bcee0fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ea19a43e23e0256d84b4ef0c82b0cfb5bcee0fda"}]},"branch":"refs/heads/master"},"61caff7c436c5982d249b9efe0857413c667e4df":{"kind":"REWORK","_number":11,"created":"2023-03-01 07:17:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/11"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 07:17:38.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61caff7c436c5982d249b9efe0857413c667e4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/61caff7c436c5982d249b9efe0857413c667e4df"}]},"branch":"refs/heads/master"},"3bf38f8403d95103f0af0928a679d8e47821a110":{"kind":"REWORK","_number":12,"created":"2023-03-01 07:47:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/875796/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/96/875796/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/96/875796/12"}}},"commit":{"parents":[{"commit":"df3fce9e6b285d846ab4945cb24de7b51a3d425e","subject":"Expose policy_default_rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/df3fce9e6b285d846ab4945cb24de7b51a3d425e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 03:02:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-01 07:47:43.000000000","tz":540},"subject":"Replace legacy facts and use fact hash","message":"Replace legacy facts and use fact hash\n\n... because the latest lint no longer allows usage of legacy facts and\ntop scope fact.\n\nChange-Id: I2a52329948becfcfb799f034847db9eae7bd903d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3bf38f8403d95103f0af0928a679d8e47821a110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3bf38f8403d95103f0af0928a679d8e47821a110"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
