)]}'
{"id":"openstack%2Fpuppet-aodh~881155","triplet_id":"openstack%2Fpuppet-aodh~stable%2Ftrain~Ide85aebb6baff65ff26f86f65994eda3a4f1451c","project":"openstack/puppet-aodh","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-21 03:12:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ide85aebb6baff65ff26f86f65994eda3a4f1451c","subject":"DNM: testing ...","status":"ABANDONED","created":"2023-04-21 02:10:16.000000000","updated":"2023-04-21 03:12:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7fa3ad590ffef5b0b96bf10aea639a92dff7846","_number":881155,"virtual_id_number":881155,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-21 02:50:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-21 02:50:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f71591318cedd1546427db7843fc72640e3f0a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-21 02:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27f7441bccdd555b77a58919d11f10ecd1881944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 02:50:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/483e1a1d8e064cd4887c618e8ebab517\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b97e514c1af54c9998018e4bd5073ad4 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd78eb7495b041a1a1a03122ac785204 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7d78e340fca44caa2e91e192b32b89f : SUCCESS in 3m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0d07902209104b8591b1501d7d377ba0 : SUCCESS in 15m 42s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/aac2d3e565de4e2aa7b8800ecdd12e1c : SUCCESS in 10m 50s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbe28b885f9942deb45ba7840b2ab05c : SUCCESS in 16m 19s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff2c39a5ffbc4c96af3117a9892293d1 : SUCCESS in 11m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/001ea5461d1340cba620a369ac455db6 : SUCCESS in 8m 25s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8a2690e419374f31a5c92171791101eb : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/96dab1ac37d847e1ae053645bbb934dc : FAILURE in 9m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0d81ce2f9cc47b88db1867fdab0ffc6 : FAILURE in 10m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7961ae23285748ba97d93de14a4cf192 : FAILURE in 11m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b9c3ae582524d18b55e8fc759565f0a : FAILURE in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7fa3ad590ffef5b0b96bf10aea639a92dff7846","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-21 03:12:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1c8f7fdba1d02e5e58032e67ebeaf46217af7c3","revisions":{"d1c8f7fdba1d02e5e58032e67ebeaf46217af7c3":{"kind":"REWORK","_number":1,"created":"2023-04-21 02:10:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/881155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/55/881155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/55/881155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/55/881155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/55/881155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/55/881155/1"}}},"commit":{"parents":[{"commit":"1179715c0f814bda2cca29c705ae59fb597c71bd","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/1179715c0f814bda2cca29c705ae59fb597c71bd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-21 01:33:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-21 01:33:55.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nChange-Id: Ide85aebb6baff65ff26f86f65994eda3a4f1451c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d1c8f7fdba1d02e5e58032e67ebeaf46217af7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d1c8f7fdba1d02e5e58032e67ebeaf46217af7c3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
