)]}'
{"id":"openstack%2Fpuppet-aodh~909809","triplet_id":"openstack%2Fpuppet-aodh~master~Ifb42750a7767579bb31f79a68f4042c8f6e8caa1","project":"openstack/puppet-aodh","branch":"master","topic":"hc-ignore_proxied_requests","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-25 16:03:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifb42750a7767579bb31f79a68f4042c8f6e8caa1","subject":"healthcheck: Expose ignore_proxied_requests parameter","status":"MERGED","created":"2024-02-22 02:25:31.000000000","updated":"2024-02-25 18:37:19.000000000","submitted":"2024-02-25 18:36:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909809-hc-ignore_proxied_requests","meta_rev_id":"24c54e38221c514951adbd4edadf49e9a35ea8dc","_number":909809,"virtual_id_number":909809,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-25 18:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-02-22 12:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-02-23 16:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-02-23 16:36:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 04:25:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 12:57:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-02-23 16:36:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d8d0960debc3d6995f24965db4a601a78c6ee49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-22 02:25:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fda609d9518b5d8519ef3857bc8902d0e215200e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-22 02:32:24.000000000","message":"Topic set to hc-ignore_proxied_requests","accounts_in_message":[],"_revision_number":1},{"id":"353510becc386627b5df4ebb3933647e741b6a7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-22 03:04:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3ad72efeb4ae440ed9869e2c0882ed4d9c35b66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 04:25:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9645aa93fdb740db9b6b67f219180def\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/547467572a524a3d845be74a482d0cd8 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec340b916a3c4a84b771e148c241dece : SUCCESS in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c95e946561d450eb6da4bf14c03979e : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bcc0e7407a147b28a965f23fa97d8d5 : SUCCESS in 10m 28s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b31acfb2ad714fccabd8a8b90dc728dd : SUCCESS in 8m 21s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/cde1547bca664d4981b83f652f7b5f52 : SUCCESS in 14m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/baf5183fc9c3429fa7e36af8e542fa6a : SUCCESS in 8m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69095200a4e0481f97d4be3fba759555 : SUCCESS in 49m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3cba0b9519745ac96439b58d84b8b10 : SUCCESS in 51m 58s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf8ef2d35cbe408ea35d5f68ce85ba11 : SUCCESS in 56m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/657da78ae4da4022ba25074a74916e45 : SUCCESS in 1h 14m 37s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/66af7e8332304f6f8c8190797de42e6f : SUCCESS in 17m 15s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dda856d20e064099a38b08827dd83502 : SUCCESS in 14m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51eb6910845f42ebb9040a72e8dbb316 : SUCCESS in 11m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26aae7a2f2bc422f88800288d2335a70 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0428cefcf955ec192c99a580080fb435b7880e3b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-22 12:57:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1974c41b5ae92a1ef365ba7485ab618db911fcb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-23 16:36:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b9d33f6370ad8e8f5ecdce57c6219933b5b44eee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:37:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dba16999f0640e0acb18f66273d985c\n\n\nWarning:\n  Change 909807 in project openstack/puppet-oslo does not share a change queue with 909809 in project openstack/puppet-aodh","accounts_in_message":[],"_revision_number":2},{"id":"174b740d1a6edff1f14ee98b5d9de231421a0195","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-24 15:19:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58057c8490c634e493c27fa93a911e8cd631169b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:24:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e93f51c4bd3e4e2e82a3f9d14c917b46\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cb9559621994c3397d5dacac78f8e06 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97439c4b878c49aa8664287c7c41a3e8 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b11cc3025f64af08b3a2c6f8ccb7738 : SUCCESS in 5m 00s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d36b8075ca044e4a7ac5d5e80aa1b4a : SUCCESS in 9m 29s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94d44f6f835b421cba8203a2abb86c80 : SUCCESS in 7m 21s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd8d70741bc440ce954bea0b69504bdc : SUCCESS in 9m 30s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/224d9c37bc2543a7a74f6c1c5696a84c : SUCCESS in 6m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70fa357808814743951264faa5c9adc8 : SUCCESS in 38m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/88b89cbcdc5e4db48b163fbb3905c163 : SUCCESS in 51m 05s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6530a72d99d74f728a0d24695a9659fc : SUCCESS in 47m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc152b068b71495fb9e0e57beb396bfb : SUCCESS in 1h 02m 42s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/06d4ebc6fed840a3bf5fbfd611daacaa : SUCCESS in 16m 18s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc1769d7908a41da9f35559e87fa286c : SUCCESS in 12m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e72bd8f36e9d43368e5a7c53a4514f62 : SUCCESS in 12m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82b616634c7f40919c62f8f819cd36ee : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e6600a9e1ffd87675eaa11fb80ab8936b1274adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:24:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5dcd35eda8d47c635dce4e7ac72c461104764a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 17:26:52.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aa3dcfa14a0426cb8df021a6de2b412\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0668eb4a6f0f4ee7a4e3b3eb483423cb : SUCCESS in 3m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3b3dedcb8bf4fbda47724cc2592450b : SUCCESS in 2m 19s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4292f756be84270b997fa7625fc0085 : SUCCESS in 11m 31s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5447fe60cd6447788c737ed0ce51656a : SUCCESS in 5m 52s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/519a1b80fa2d4e08ba2842b5125952b9 : SUCCESS in 54m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32e6f797863f45a082190c4da8c11656 : FAILURE in 1h 01m 30s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/d928bd2abc7c4eb698e0b1c99a7eb1d7 : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/976f193d09ec474fbcc47a5b34d2a8a4 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":2},{"id":"b4b97173fde8bcc98f455a977feeed3ef6047565","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-25 16:03:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a7593feba857949d1c9caafc00d7d7ede45bfa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 17:34:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013a09e7219a4a298ce2e1461ffb04d7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a96fe6d567ab4282a7e3a862adccad98 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89e4e66331b246e79b19099067016dc8 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3965c818219e46e19d39b957fe42f0ff : SUCCESS in 6m 11s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ef3d99078664118ade67da9fccec2dc : SUCCESS in 11m 39s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a92d6e5bec0749a8b45acd1731929a03 : SUCCESS in 6m 22s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbf15f3df1724c48a2f1267c2a30c058 : SUCCESS in 11m 07s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c0212878a9248a28fc91befc98bdecc : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf7d451bbf704d1e941a0772e84819cb : SUCCESS in 38m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d8df293fd154d7e89e2172843a1ac90 : SUCCESS in 51m 28s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6d8ca5f8c994e10bb219beb8b8c1578 : SUCCESS in 55m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74f2f62c5a484a139870d743b47a54f3 : SUCCESS in 1h 24m 13s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c4be5dc1d494ef680edcdd8d85eb7ed : SUCCESS in 19m 15s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/862be644f26d43ad9ded719dda63e191 : SUCCESS in 16m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4212a569462c441d9b65f0f1d413203b : SUCCESS in 9m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/551ae74f330f4d3a8b5a5c77a98f3db2 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"17ba89a3239a374f79122d9f9366d0a5fe355269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 17:34:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0760c4d4653fc1b3930c5cf02bedc699e297d400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 18:36:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/914e88c9cc814d27ad19b4ab4cac841e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ec7818b07604da9b59e44b9e1ff7db9 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/401ea6cf913b4b0e9a366585ecca6592 : SUCCESS in 2m 25s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/123fcf619a934bdf8c5b1361eff9dfe7 : SUCCESS in 8m 11s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f66058bade54f16bb9e3a8146278816 : SUCCESS in 6m 16s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3896370f46f49ffb84e67bd2a0b03ba : SUCCESS in 46m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c9e5c6bb73a41cdbac16b8f02b04fef : SUCCESS in 1h 00m 59s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/63ff0a04cbcf4689829b9b30ca7e375f : SUCCESS in 18m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df9bbda8bf0841cdb3acd908ac872110 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":2},{"id":"ccc3f1b91bc5d9826050389e3c0ef692e7113007","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 18:36:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24c54e38221c514951adbd4edadf49e9a35ea8dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 18:37:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88d147a412c742208138cff20d3aad01\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c331f6093114e3d8d77d13fc585f00d : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9d14c5f8e1a1ea6f35cca4687997e76ee09805c0","revisions":{"17bcaa23cd57e966a58721f10c3dac6e63c105c0":{"kind":"REWORK","_number":1,"created":"2024-02-22 02:25:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/909809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/09/909809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/1"}}},"commit":{"parents":[{"commit":"7031f2a360b8c30b9eca07bd1d291586b00f66ff","subject":"Merge \"reno: Update master for unmaintained/yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7031f2a360b8c30b9eca07bd1d291586b00f66ff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-22 02:24:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-22 02:24:54.000000000","tz":540},"subject":"healthcheck: Expose ignore_proxied_requests parameter","message":"healthcheck: Expose ignore_proxied_requests parameter\n\nDepeneds-on: https://review.opendev.org/909807\nChange-Id: Ifb42750a7767579bb31f79a68f4042c8f6e8caa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/17bcaa23cd57e966a58721f10c3dac6e63c105c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/17bcaa23cd57e966a58721f10c3dac6e63c105c0"}]},"branch":"refs/heads/master"},"9d14c5f8e1a1ea6f35cca4687997e76ee09805c0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-22 03:04:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/909809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/09/909809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/09/909809/2"}}},"commit":{"parents":[{"commit":"7031f2a360b8c30b9eca07bd1d291586b00f66ff","subject":"Merge \"reno: Update master for unmaintained/yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7031f2a360b8c30b9eca07bd1d291586b00f66ff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-22 02:24:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-22 03:04:40.000000000","tz":0},"subject":"healthcheck: Expose ignore_proxied_requests parameter","message":"healthcheck: Expose ignore_proxied_requests parameter\n\nDepends-on: https://review.opendev.org/909807\nChange-Id: Ifb42750a7767579bb31f79a68f4042c8f6e8caa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9d14c5f8e1a1ea6f35cca4687997e76ee09805c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/9d14c5f8e1a1ea6f35cca4687997e76ee09805c0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
