)]}'
{"id":"openstack%2Fpuppet-aodh~928879","triplet_id":"openstack%2Fpuppet-aodh~master~I831108b202998dc02658c4b1ec2ef6e300516627","project":"openstack/puppet-aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-11 21:58:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I831108b202998dc02658c4b1ec2ef6e300516627","subject":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout","status":"MERGED","created":"2024-09-11 02:43:02.000000000","updated":"2024-09-11 21:59:45.000000000","submitted":"2024-09-11 21:58:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928879","meta_rev_id":"b100140eb9bcc5cd8edc725d796f22836aa87a8c","_number":928879,"virtual_id_number":928879,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-11 21:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-09-11 09:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-09-11 20:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-09-11 20:37:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-11 03:02:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-11 09:46:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-09-11 20:37:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"85f7db25489d71f9b62b7b2524b28ca5991c959c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-11 02:43:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4212c98bf9c1fe40b2bb773b6a5509e7bb084731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 03:02:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df09c47ea35444af9a233bb0bc11be09\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/745e57fc3d894247b8971c14f9ecf8a3 : FAILURE in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5521ea7825d44573aecbc675ebc74c30 : FAILURE in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4660840c43424a32a8b2893db1ac5223 : FAILURE in 5m 33s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4387fbdf10f44c67bd771699e016638d : FAILURE in 7m 14s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52d0fdc9c2444011be408eaae868128a : FAILURE in 5m 37s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f2318a1d4fb4c10a399c446f5f7c92c : FAILURE in 7m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/351b260143e949799fc3ac4d73dc8ce2 : FAILURE in 5m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d80a08d82d914895b3ea166e87135d41 : FAILURE in 5m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fd7aff2e4fd4e84ab8210e8ce5b7887 : FAILURE in 9m 55s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30485b2a7b9c40aca344201309a58ed4 : FAILURE in 5m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f6ff92a7a8841d69e4e703ae1517aad : FAILURE in 7m 10s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/824b4078e9b44c69bcdfb1d91529c6f1 : FAILURE in 6m 32s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/979fd662ed9249a9ad3be7c19709f8e8 : FAILURE in 5m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c688063e64564f6d95b0f1e081bf307c : FAILURE in 3m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9375fc90bed4991900cb6ee6f041a64 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3d22f6834a8074c1271c7e7305d7d545c7f78955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-11 09:21:43.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb26ffd5ee0f0fcfea1437cfdab300b59226b294","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-11 09:46:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c853e44dbee6441a335f5f5bab2a0d88ea825d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 10:15:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3874e41f76e5436fb7ff97ade340805e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dcd361d85774d8e8fdbf7ca29bee44b : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf5a641b9e1f44c9b6dbae8df346fb03 : SUCCESS in 2m 03s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c374ffe72c064e119410e78274da9e60 : SUCCESS in 6m 03s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f93402ea77e4d62ba88f3e5ea32cf27 : SUCCESS in 9m 17s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5b51df62b764b1d999f47a111993dff : SUCCESS in 4m 02s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/928ff8fa63c74649b6968f6e6a21c22e : SUCCESS in 8m 40s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f30e07c3f9694ffbab67b107bc572751 : SUCCESS in 6m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0cf98fb22bf4e28918e522444215369 : SUCCESS in 29m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c9d5f9557e848e68c75721180ca01ae : FAILURE in 51m 56s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8635b51d56664c28ab8fd81a9bdbcbaa : SUCCESS in 48m 55s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/beaef7490ddd49d08d45af1d0211c016 : FAILURE in 50m 41s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e38863c0b9c64f5fb4fa21f35e5c7e75 : SUCCESS in 19m 58s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d42d936253846928fa7716517299261 : SUCCESS in 10m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7efe1acccaab40c9b2330a3665bcf455 : SUCCESS in 12m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bbfc207b8564b078a34af6e3a9062b4 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f5cf3691ef3f17454f0568ba320750255d34f7ed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-11 13:03:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a75971a126ee825b092849890b2b36d57124cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 14:08:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70355150b7e4806afbbc4f5055fc829\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c7955dc113140c09f4948676abc1a91 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a4f0523cb4d4e8795ee4c1305ba722b : SUCCESS in 2m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/426a0e79b3b6456eba3d8e2ea57ee2bd : SUCCESS in 4m 15s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/195fd1087aeb479fac9aa35bb60cfea2 : SUCCESS in 7m 34s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23a5ff18b6a04d2684b0a76f8c360b10 : SUCCESS in 5m 24s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b09823e846ad43428d93c9c4df1a71fb : SUCCESS in 5m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9b5a0952ada4483a3dd1b8270b9242c : SUCCESS in 4m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31f73d69e3684f639813e756134a1337 : SUCCESS in 28m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6237df585d740ec8a21042b8ed7bf24 : SUCCESS in 57m 43s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8746e90f64c24060bedadfa668cecc97 : SUCCESS in 1h 03m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a286526dc9045749ff48a9bd4f98557 : SUCCESS in 47m 44s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/2121b8f07e0b47a89618284866c8a8dc : SUCCESS in 19m 14s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97b02613cca24d79a3ef4f49e7256364 : SUCCESS in 10m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbe39d67af5443caa4e8f5b38db430e6 : SUCCESS in 10m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/871a6bd30dfe4b96b04bf7d24f6df9cd : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":2},{"id":"a31e63705ca00ae74934d0b4460f14d0329a0666","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-09-11 20:37:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9f87f98e7b0e9cb52c0deb37fdf4f8ba683b46e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 20:38:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"642c569df20ca72457c38eb004a2c6933e40ef08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:58:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d5f448c60844cf19ad609a65c82c355\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9294605ecb44632b83bd275ec04c111 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7bbff44ee4149448f6e5970c640d1f0 : SUCCESS in 2m 25s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb69b5c1418b4853a3eb11e97bb6f52a : SUCCESS in 7m 30s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce1f214e08534f1fbb438158a6bf7ca8 : SUCCESS in 5m 48s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1f6ed50a9bc43b889f8441d4169b24b : SUCCESS in 52m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80752657535c4067b7cd50b09c947abb : SUCCESS in 1h 14m 53s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/8714b9a6396448329b676dfb33c6ab16 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a28a90e173c48ebb30722cf75fe825b : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1c48a16b76dc563a53881820d8c10856e4117634","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:58:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b100140eb9bcc5cd8edc725d796f22836aa87a8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:59:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a82c962484f4872970f6a98ebabafcd\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ede37ef9731f4785ba27ea62504f7e71 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba7a10c3cd394746e031de4d3aab01ab72d8e72f","revisions":{"a6089b6b49571403882757d305e173df60a099d0":{"kind":"REWORK","_number":1,"created":"2024-09-11 02:43:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/928879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/79/928879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/1"}}},"commit":{"parents":[{"commit":"96bd6ba987b8925feea98ca66b1029311ee6ce5a","subject":"rabbitmq: Support queue manager options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/96bd6ba987b8925feea98ca66b1029311ee6ce5a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 02:42:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 02:42:29.000000000","tz":540},"subject":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout","message":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout\n\nDepends-on: https://review.opendev.org/928497\nChange-Id: I831108b202998dc02658c4b1ec2ef6e300516627\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a6089b6b49571403882757d305e173df60a099d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/a6089b6b49571403882757d305e173df60a099d0"}]},"branch":"refs/heads/master"},"ba7a10c3cd394746e031de4d3aab01ab72d8e72f":{"kind":"REWORK","_number":2,"created":"2024-09-11 09:21:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/928879/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/79/928879/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/79/928879/2"}}},"commit":{"parents":[{"commit":"96bd6ba987b8925feea98ca66b1029311ee6ce5a","subject":"rabbitmq: Support queue manager options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/96bd6ba987b8925feea98ca66b1029311ee6ce5a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 02:42:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-11 09:21:43.000000000","tz":0},"subject":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout","message":"Add support for [oslo_messaging_rabbit] rabbit_stream_fanout\n\nDepends-on: https://review.opendev.org/928497\nChange-Id: I831108b202998dc02658c4b1ec2ef6e300516627\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ba7a10c3cd394746e031de4d3aab01ab72d8e72f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ba7a10c3cd394746e031de4d3aab01ab72d8e72f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
