)]}'
{"id":"openstack%2Fpuppet-aodh~933097","triplet_id":"openstack%2Fpuppet-aodh~master~I0353a41cb079607853b017f56f5b2e66bce3bfe5","project":"openstack/puppet-aodh","branch":"master","topic":"dev_access_log_env_var","attention_set":{},"removed_from_attention_set":{"7888":{"account":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"last_update":"2024-10-23 16:55:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0353a41cb079607853b017f56f5b2e66bce3bfe5","subject":"add access_log_env_var parameter to aodh::wsgi::apache","status":"MERGED","created":"2024-10-23 12:04:51.000000000","updated":"2024-10-23 16:56:56.000000000","submitted":"2024-10-23 16:55:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933097-dev_access_log_env_var","meta_rev_id":"dcfa46f154441cf99f10ee9859f87a0b380a8e2f","_number":933097,"virtual_id_number":933097,"owner":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-23 16:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-23 15:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-10-23 14:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-23 15:34:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-23 13:06:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-23 14:32:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-23 15:34:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"bfce3d922d2905dd1c4dbac9f7184153d953fc5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2024-10-23 12:04:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a037953314ae1fd91cce1e7bd69a9ccc26adebd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 13:06:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ead8624df9541fdb03bbd65b3748a9e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d7fe39a62f5432ca8fde45832fc2079 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86b967c0ebeb4e07a5fa5f95de01b480 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfd18d7dd2404a56a00721691be52052 : SUCCESS in 5m 01s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/148a15b39ed748ecb9acd759c8e2033e : SUCCESS in 6m 48s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86ecb7708b74e84b3d40d6d48c84ad1 : SUCCESS in 5m 22s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b4a06e4e9294fd4acae8dc61927a1d7 : SUCCESS in 8m 53s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b662823c07d9490ba6350d8b5261586e : SUCCESS in 4m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4441186000ca44478f16a3603bbba5c8 : SUCCESS in 35m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aefa7ae149864e4d912b33cd18920c83 : SUCCESS in 46m 24s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d58272a03235432dbc723ce0e4ea8532 : SUCCESS in 55m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfc5edb627124bf08144e4c1d2ef18ee : SUCCESS in 58m 07s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/694c74934ca044da84bbd15eb479b32d : SUCCESS in 14m 45s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cba142752a947f3bc61b0c228e2e5a7 : SUCCESS in 10m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0db61eb7df41412983704f75ee6b9787 : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae098415495ab8f89202b19617a8ce09eec1fce7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2024-10-23 13:22:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"60e519ea25262f1e9efe48f3f0ee8e0ebe5b1cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:29:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eedd5bee05044e8f96da54f0d4585217\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/164dd2e379c94bdbaac83aa7dfca084a : SUCCESS in 3m 13s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aedd5524e954469481a5badec3309d0f : SUCCESS in 3m 11s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/44a973bc532d4ab2a953b5d0c29e1271 : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e154bcf949064428bc00eea27d25c4a5 : SUCCESS in 9m 51s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b581700ec844597b4d3324b532ee7b7 : SUCCESS in 4m 16s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0257f9f500846bcb35e09c2340e4e34 : SUCCESS in 7m 21s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0e214f4f6ae4662bf34d5e8efb2a008 : SUCCESS in 5m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfc6ca612c944c08bbc19cf6dc9ef1f2 : SUCCESS in 27m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c5616d78b884b48b2da14b95ab34c54 : SUCCESS in 59m 23s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3615a7b9c6e428ca2953137cded427b : SUCCESS in 45m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28e1be48d7244ade9785d647366da41e : SUCCESS in 48m 14s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bc318f337d44067a128d157d3c0d739 : SUCCESS in 15m 37s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aafbecf9c1554f0c832468533817619c : SUCCESS in 12m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69a47383faee439d9437159787e0e11e : SUCCESS in 9m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a46222ba71034201b95a252f2dfb00c7 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d2af151961fe664df804631ab3cf708146f18799","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-23 14:32:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cb5a8a76cb48e000e89d7ffda77d80aaffbae836","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 15:34:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"77407ab03216f477d646d5bf3b90d5d3a099ca58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 15:35:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"635ddba62b2e621c1e0bcf2be9ee8584daf85537","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:55:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da6187b3da614f5daa40fdf83ff9c159\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f1129b48ef343b2aa188b6e80821b04 : SUCCESS in 2m 01s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a2b41d725324d9ab8846c0216f09260 : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/14b0a6a86d1a489481609f37aa947195 : SUCCESS in 5m 57s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b1820f0e1f648f587ac79e316128aba : SUCCESS in 5m 51s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07ef07147ee140cd8afff3ea208d2600 : SUCCESS in 37m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f49e512cd6fe447f97ce1998f3530791 : SUCCESS in 1h 19m 30s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/bab2682f9dbf4ff7a00e0bec4e2b25e4 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e60da3e924c9461a861cdf138e17f6bc : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d54d88cde0b97dc2aa7a5ee5b8506e55b3075471","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:55:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dcfa46f154441cf99f10ee9859f87a0b380a8e2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:56:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb93ff864094ed287848a2bdf794361\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c25dd7c48cc74b888e78aa17f9841ec7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8aa672adeab547da159fdbb50b695fb415dd583","revisions":{"63c314fc81e8c179afe4991529d5631b01f3dc44":{"kind":"REWORK","_number":1,"created":"2024-10-23 12:04:51.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/97/933097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/97/933097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/1"}}},"commit":{"parents":[{"commit":"0a8e8a7f939f09342b6efd6a4763addca6bde702","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0a8e8a7f939f09342b6efd6a4763addca6bde702"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2024-10-22 19:08:44.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2024-10-23 12:04:48.000000000","tz":120},"subject":"add access_log_env_var parameter to aodh::wsgi::apache","message":"add access_log_env_var parameter to aodh::wsgi::apache\n\nChange-Id: I0353a41cb079607853b017f56f5b2e66bce3bfe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/63c314fc81e8c179afe4991529d5631b01f3dc44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/63c314fc81e8c179afe4991529d5631b01f3dc44"}]},"branch":"refs/heads/master"},"b8aa672adeab547da159fdbb50b695fb415dd583":{"kind":"REWORK","_number":2,"created":"2024-10-23 13:22:33.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/97/933097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/97/933097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/97/933097/2"}}},"commit":{"parents":[{"commit":"0a8e8a7f939f09342b6efd6a4763addca6bde702","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0a8e8a7f939f09342b6efd6a4763addca6bde702"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2024-10-22 19:08:44.000000000","tz":120},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2024-10-23 13:22:12.000000000","tz":120},"subject":"add access_log_env_var parameter to aodh::wsgi::apache","message":"add access_log_env_var parameter to aodh::wsgi::apache\n\nChange-Id: I0353a41cb079607853b017f56f5b2e66bce3bfe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b8aa672adeab547da159fdbb50b695fb415dd583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/b8aa672adeab547da159fdbb50b695fb415dd583"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
