)]}'
{"id":"openstack%2Fpuppet-aodh~942028","triplet_id":"openstack%2Fpuppet-aodh~master~I83f663f5378e20cbb40f3c1d903f74634dfcd81e","project":"openstack/puppet-aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-19 16:07:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83f663f5378e20cbb40f3c1d903f74634dfcd81e","subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","status":"MERGED","created":"2025-02-18 01:35:00.000000000","updated":"2025-02-19 16:08:53.000000000","submitted":"2025-02-19 16:07:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942028","meta_rev_id":"24156a5afc9af3ea45d4052ca83b31fb9a2f49aa","_number":942028,"virtual_id_number":942028,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-19 16:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-02-18 15:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-02-19 14:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-02-19 14:11:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-18 04:43:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 15:04:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-02-19 14:11:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"362a31d40b8b18ac55c20a91022cc7bc77c171e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 01:35:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd43fd85c28dd5585dad5aab49abbd30de9f3cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:17:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dda89b7214ed86ca3e49448a787dbe1f7c63697f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:53:25.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c396bf0148dc7b6335514d54feaad8d6b8583010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 04:43:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0aa8396a1e43c58b6725e19c8468ff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf8721e4a80e4f709ece46481797e77e : SUCCESS in 2m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/708dccd585a74f0dbf18a1ac7e8026aa : SUCCESS in 3m 12s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c31d5c2743504550aea847726add1418 : SUCCESS in 6m 22s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba34aaa856224b54a185e71833116646 : SUCCESS in 11m 21s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd9e25ce7fe54d8f8675faf606cc1bad : SUCCESS in 8m 38s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/97b40dcff67047eeb16edafcd05d9441 : SUCCESS in 15m 06s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a33e5e2dc0a452690cb05f4392e7eda : SUCCESS in 6m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3dd947d4632422d839532baafe7bafd : SUCCESS in 48m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/faa76cca58854873a4bbd6454e6aef8b : SUCCESS in 1h 20m 22s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95d36b4b133042dd8905f837e82f56c4 : SUCCESS in 1h 37m 41s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c44c6e56c6554068a13d200f60dc7fe6 : SUCCESS in 59m 56s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8ec340976934607b6cd1587990951e0 : SUCCESS in 1h 31m 07s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51a7d7e4b4904b02b91936dfda1abe8d : SUCCESS in 1h 36m 58s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5580ecffb084b6abbde5448cc7f1722 : SUCCESS in 29m 31s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d140c19ab61e4ee3b03bb1711b1f5618 : SUCCESS in 11m 34s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/53008f56b2ae4747b66251f3a32063d8 : SUCCESS in 31m 15s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ac0a2a141f84631a8ea1909ada55419 : SUCCESS in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/157285a1073444c6a7f65b583da16a94 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"2c188bff6611724eca154f40d9ec163be044a4f4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-18 15:04:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a5bfb8ef3763cd385c4b6324caf6651114601350","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-19 14:11:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aca56d4bdd647cd32c5b8de4f314125cdc27eef5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 14:13:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"57cb4025ccb9cd655e2ec8b110e0154ce9b0db03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:07:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026d3e70437544368bfb1ebc4d130aca\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4216b3a48dc47e6a91cc379bf9bca0f : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f15492854214a1b80ab9086350393ac : SUCCESS in 2m 24s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/421e17c6b13e421cb78fde1c330efc3d : SUCCESS in 10m 46s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ffbaf127e2f4825bfb620f552fb36bb : SUCCESS in 5m 59s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfdce11beb5b4b9ab76ea13f83772f8a : SUCCESS in 9m 26s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a2f7be961454e9382039e8319c1c4ec : SUCCESS in 1h 49m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bb3ad7145e54acab16523e8d5e52e52 : SUCCESS in 1h 45m 08s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7c0171556d64d88929ad41464e3f2d7 : SUCCESS in 39m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc6999f9470546b4b4782ca15cd46dd6 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":3},{"id":"0de537cc1b6ad7009e79ec280cf3fa98dac36b79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:07:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"24156a5afc9af3ea45d4052ca83b31fb9a2f49aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:08:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024f2fd52760469a9e840172a85f0f0e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b208c00132a441678060e67fae530619 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"825739a1d5130be7c670fb8b8cdf14c03bc84911","revisions":{"4fb2a9e37186d4d927a5df644cceb2a4dd0d57cf":{"kind":"REWORK","_number":1,"created":"2025-02-18 01:35:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/942028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/28/942028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/1"}}},"commit":{"parents":[{"commit":"ff8c302be06f66d55663b6cd07239c1ab27bb812","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff8c302be06f66d55663b6cd07239c1ab27bb812"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 01:32:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 01:34:38.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: I83f663f5378e20cbb40f3c1d903f74634dfcd81e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4fb2a9e37186d4d927a5df644cceb2a4dd0d57cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4fb2a9e37186d4d927a5df644cceb2a4dd0d57cf"}]},"branch":"refs/heads/master"},"107204053560edc25965bc1d2a36c14e99d8b3ce":{"kind":"REWORK","_number":2,"created":"2025-02-18 02:17:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/942028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/28/942028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/2"}}},"commit":{"parents":[{"commit":"ff8c302be06f66d55663b6cd07239c1ab27bb812","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff8c302be06f66d55663b6cd07239c1ab27bb812"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 01:32:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:16:58.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: I83f663f5378e20cbb40f3c1d903f74634dfcd81e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/107204053560edc25965bc1d2a36c14e99d8b3ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/107204053560edc25965bc1d2a36c14e99d8b3ce"}]},"branch":"refs/heads/master"},"825739a1d5130be7c670fb8b8cdf14c03bc84911":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-18 02:53:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/942028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/28/942028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/28/942028/3"}}},"commit":{"parents":[{"commit":"ff8c302be06f66d55663b6cd07239c1ab27bb812","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ff8c302be06f66d55663b6cd07239c1ab27bb812"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 01:32:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:53:05.000000000","tz":540},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: I83f663f5378e20cbb40f3c1d903f74634dfcd81e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/825739a1d5130be7c670fb8b8cdf14c03bc84911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/825739a1d5130be7c670fb8b8cdf14c03bc84911"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
