)]}'
{"id":"openstack%2Fpuppet-aodh~953014","triplet_id":"openstack%2Fpuppet-aodh~master~I19aaa537a05b605f297c4fd317254c84958eae4a","project":"openstack/puppet-aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-06-21 22:38:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I19aaa537a05b605f297c4fd317254c84958eae4a","subject":"Add stream_fanout","status":"ABANDONED","created":"2025-06-20 20:20:24.000000000","updated":"2025-06-21 22:38:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0a565a603a5475fc4316f7d10c3c15e39b182d0c","_number":953014,"virtual_id_number":953014,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-21 21:14:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-20 20:51:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60efcdd1f13da06153d6a0249df9d2879e2c1716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-20 20:20:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69393cd4f019395c9d58c917012154a79720a4c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-20 20:31:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"16763939665592aecdb529eb05b3f7b6b574a517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 20:51:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97180b89106347d1b08a13afd98f8f4e\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1ab65fd34824dcfac6bb019892a79d6 : FAILURE in 2m 15s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9922e7cc56ba41eb88cb92ca872ae32b : FAILURE in 2m 24s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1073d129ec14f3da14da1065cc090d7 : FAILURE in 2m 25s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/25ad5559cc684ad0a826e6e3925163b8 : FAILURE in 8m 45s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fc29d966cee4e84a274f7efc011bcf3 : FAILURE in 4m 04s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b04901c7f2674c5cad7cfee536736e30 : FAILURE in 4m 15s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ce7679576ea44489415b725ab2242e0 : FAILURE in 3m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f2099ef25b54d5ab9b8ed5cfcfcaa3d : FAILURE in 10m 17s\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7bf1ddf0eba412fb7263a2770fd0c26 : FAILURE in 4m 06s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/176809207c314ff081e1ddac8134ef35 : FAILURE in 9m 05s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/e718016620964782bb909c3226283045 : FAILURE in 7m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09c74579a80f4e1d9456694d8fcb2acd : FAILURE in 4m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/408da1a989d243eb924cdaf17b37182a : FAILURE in 3m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bb2c903976e42d694200b2bd5252d5e : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a75a499c07f6bf5a4594c1dd1b500a7476a37792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-20 23:10:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"acad485e6800ac2cae2b0993dc48955a1d620718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 23:25:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45ed1ea105de4e0da890a14026a5188c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d531aca4718454aa6b0bc0ac55f2b20 : FAILURE in 4m 51s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9aa8f78aeb634268a3f0dfa9c0d76100 : FAILURE in 2m 13s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bf88eea650641568c51a6c3d2208b04 : FAILURE in 2m 10s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/53b32789a1984eadbd53fde642cdc8f7 : FAILURE in 8m 16s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b80327d8f4454c8886d1df290937e68b : FAILURE in 3m 50s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f9d033d8b584c358d7eea37a347f49e : FAILURE in 5m 24s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44d86e5198124172b3fbf505b374010c : FAILURE in 5m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fe683c157a04b658674328e054d2aeb : FAILURE in 8m 54s\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f040b1e8d70945fba9d47a8ba20c4d64 : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79ad652683104adf8cfd9cae70be989b : FAILURE in 7m 09s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c547ae0e4204421b8513603d1e8db27 : FAILURE in 7m 32s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc3c12af12c54bd183be82a4937935f5 : FAILURE in 3m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40369d73b01245c8abf7d203674425af : FAILURE in 3m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/521dd60c7ef847c1ba6391e9da3053ca : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ac102fe8607878a291df97a292559a34d640a00e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-21 20:58:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b6b97ec98a15aac6dad89e79623fec6a66ed1ffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-21 21:01:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a3434ddcf8d3b8be9ef92b48804188a463190dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 21:14:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ac46ec9ada241a2ae875017251bd059\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5393027a784444e289593d771e101aa5 : FAILURE in 2m 08s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcd90440af6242c18c5e4628570da0f2 : FAILURE in 2m 10s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b4629f8738843768b223a8ce959c30d : FAILURE in 1m 59s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/d70b8a900f4449b8ac5cddf6f436940b : FAILURE in 7m 16s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f9f9d99f3f14f4780a9d5c3acf893ef : FAILURE in 5m 13s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6c8118cd3124d1ea0f63c4004130eee : FAILURE in 3m 40s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/562119da28a3454190b88bad05304140 : FAILURE in 3m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2be1676577ce43d4a1cb5edd25902e37 : FAILURE in 9m 48s\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf6fc71249c147d59b4d183bfb3dcec1 : FAILURE in 4m 38s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/993df49d7cee4a90baad205da994f9da : FAILURE in 7m 17s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/053a4c00bd4a42a085e02fb651b46d9b : FAILURE in 6m 43s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55bc00fbf14c4ff7bbd0099b4a5cc8c7 : FAILURE in 3m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/174915df3313425d99c4c8e85a3bb34e : FAILURE in 3m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f762c3136a134acfbf17836f62b670d3 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":5},{"id":"0a565a603a5475fc4316f7d10c3c15e39b182d0c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-06-21 22:38:46.000000000","message":"Abandoned\n\nalready there","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5b73381389b784160f52d04247bf5924a7901d58","revisions":{"43debede612a275ed41fc1b5132aa07e372c2e93":{"kind":"REWORK","_number":1,"created":"2025-06-20 20:20:24.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/953014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/14/953014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/1"}}},"commit":{"parents":[{"commit":"6083fe163d6933d2fa86039f74bae4955c70416e","subject":"Remove watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6083fe163d6933d2fa86039f74bae4955c70416e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 20:19:38.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 20:20:20.000000000","tz":120},"subject":"Add queue_manager and stream_fanout","message":"Add queue_manager and stream_fanout\n\nThis patch adds two new parameters to the toplevel class:\n- rabbit_use_queue_manager\n- rabbit_stream_fanout\n\nthat are already handled by the oslo::messaging::rabbit class, which\nmakes it impossible to set using keystone_config.\n\nChange-Id: I19aaa537a05b605f297c4fd317254c84958eae4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/43debede612a275ed41fc1b5132aa07e372c2e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/43debede612a275ed41fc1b5132aa07e372c2e93"}]},"branch":"refs/heads/master"},"57476b34b9a701b9656836f7ab2828d47570cd0d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-20 20:31:24.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/953014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/14/953014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/2"}}},"commit":{"parents":[{"commit":"6083fe163d6933d2fa86039f74bae4955c70416e","subject":"Remove watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6083fe163d6933d2fa86039f74bae4955c70416e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 20:19:38.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 20:31:10.000000000","tz":120},"subject":"Add queue_manager and stream_fanout","message":"Add queue_manager and stream_fanout\n\nThis patch adds two new parameters to the toplevel class:\n- rabbit_use_queue_manager\n- rabbit_stream_fanout\n\nthat are already handled by the oslo::messaging::rabbit class, which\nmakes it impossible to set using aodh_config.\n\nChange-Id: I19aaa537a05b605f297c4fd317254c84958eae4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/57476b34b9a701b9656836f7ab2828d47570cd0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/57476b34b9a701b9656836f7ab2828d47570cd0d"}]},"branch":"refs/heads/master"},"ce3d5488a2ea5b06000cf3dc2543e0c409f46140":{"kind":"REWORK","_number":3,"created":"2025-06-20 23:10:53.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/953014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/14/953014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/3"}}},"commit":{"parents":[{"commit":"6083fe163d6933d2fa86039f74bae4955c70416e","subject":"Remove watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6083fe163d6933d2fa86039f74bae4955c70416e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 20:19:38.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 23:10:42.000000000","tz":120},"subject":"Add queue_manager and stream_fanout","message":"Add queue_manager and stream_fanout\n\nThis patch adds two new parameters to the toplevel class:\n- rabbit_use_queue_manager\n- rabbit_stream_fanout\n\nthat are already handled by the oslo::messaging::rabbit class, which\nmakes it impossible to set using aodh_config.\n\nChange-Id: I19aaa537a05b605f297c4fd317254c84958eae4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ce3d5488a2ea5b06000cf3dc2543e0c409f46140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/ce3d5488a2ea5b06000cf3dc2543e0c409f46140"}]},"branch":"refs/heads/master"},"412dc4d08a125ec9b080ce6f4aa7b347c070bd39":{"kind":"REWORK","_number":4,"created":"2025-06-21 20:58:16.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/953014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/14/953014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/4"}}},"commit":{"parents":[{"commit":"6083fe163d6933d2fa86039f74bae4955c70416e","subject":"Remove watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6083fe163d6933d2fa86039f74bae4955c70416e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 20:19:38.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-21 20:57:28.000000000","tz":120},"subject":"Add queue_manager and stream_fanout","message":"Add queue_manager and stream_fanout\n\nThis patch adds two new parameters to the toplevel class:\n- rabbit_use_queue_manager\n- rabbit_stream_fanout\n\nthat are already handled by the oslo::messaging::rabbit class, which\nmakes it impossible to set using aodh_config.\n\nChange-Id: I19aaa537a05b605f297c4fd317254c84958eae4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/412dc4d08a125ec9b080ce6f4aa7b347c070bd39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/412dc4d08a125ec9b080ce6f4aa7b347c070bd39"}]},"branch":"refs/heads/master"},"5b73381389b784160f52d04247bf5924a7901d58":{"kind":"REWORK","_number":5,"created":"2025-06-21 21:01:35.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/953014/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/14/953014/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/14/953014/5"}}},"commit":{"parents":[{"commit":"6083fe163d6933d2fa86039f74bae4955c70416e","subject":"Remove watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/6083fe163d6933d2fa86039f74bae4955c70416e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-20 20:19:38.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-06-21 21:00:54.000000000","tz":120},"subject":"Add stream_fanout","message":"Add stream_fanout\n\nThis patch adds a new parameters to the toplevel class:\n- rabbit_stream_fanout\n\nthat is already handled by the oslo::messaging::rabbit class, which\nmakes it impossible to set using aodh_config.\n\nChange-Id: I19aaa537a05b605f297c4fd317254c84958eae4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5b73381389b784160f52d04247bf5924a7901d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5b73381389b784160f52d04247bf5924a7901d58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
