)]}'
{"id":"openstack%2Fpuppet-aodh~958140","triplet_id":"openstack%2Fpuppet-aodh~master~I179246eabd540e7c905d67eb3f99840809c99e4e","project":"openstack/puppet-aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-29 15:26:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I179246eabd540e7c905d67eb3f99840809c99e4e","subject":"Refactor api service management","status":"MERGED","created":"2025-08-21 05:13:19.000000000","updated":"2025-09-29 15:26:27.000000000","submitted":"2025-09-29 15:26:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958140","meta_rev_id":"9c232b347fc0ced6695ecd3ba52fa39bf875c578","_number":958140,"virtual_id_number":958140,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-29 15:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-09-29 06:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-09-29 14:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-09-29 14:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 05:43:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 06:18:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-09-29 14:17:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"8730721629a0a122a57745a4ff6146e702afe521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 05:13:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"602490a0168aa89f1ba672846b4a54d415f818a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 05:25:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9caf4c3023831ff2db21ea78e727efe998b466ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 05:43:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdf21ed65dbe49639e3f6fa3f6eee8f0\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b69498cdb339410bbfb95fe6941be451 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36f5c4161fb6475888240a4c56e97807 : FAILURE in 2m 34s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/664919e0678743a28032e71bfca55231 : FAILURE in 6m 03s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/688ab9a41bb24227bd9d1e1180a8634b : FAILURE in 3m 09s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e3341ac802c4f05a51a845ce2c2fdb5 : FAILURE in 4m 46s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32af9f05e6ae4726b08341853721fc62 : FAILURE in 5m 30s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebc3454a07c4469cb6201a7bd3d8a1cd : FAILURE in 10m 43s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/946ceb2ad7b548b5920434c55e4db8e0 : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b94350cbdab04fd48cd4e6a252821353 : FAILURE in 5m 04s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ec6aed6aed74756a9420437150d255d : FAILURE in 9m 20s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbd9aab1a4f44ba9a643df068c411316 : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/efaff018b6d94ddd98aa2cf16c94731f : FAILURE in 7m 01s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2dce05347ef49d8a03224d3cdc496b5 : FAILURE in 4m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0ac920c53b94681a6fca4037e722d09 : FAILURE in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6521950bf46041506c65b7daf99237c64b52fac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 05:45:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d6ef57b4ba5bc08fc71a8337eb82a8dc00c32255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 07:06:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8b6ee59eba878b620e6a4b6b43ce27b3937b7bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 07:15:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"30634c3e27e65d1cbda1907e71002dddfd3451fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 08:47:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5f05c36fd194f979f210575926c5bab\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04a0245ea3464067bbaa225e45a6fb44 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10f9aaa123d7467ead5f2fb17e34d091 : SUCCESS in 3m 09s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/50fa47dc3f26403a8bb4be3d21dabfa8 : FAILURE in 9m 08s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63795f1350e941b9a76a58f50d937f4d : FAILURE in 3m 00s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34b013e9becd4dafba0e6f683f6a71de : FAILURE in 6m 37s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/485b26b459e444f296e7c386be27f972 : SUCCESS in 29m 02s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d421a44bc3c41939d720f5cab9d1906 : SUCCESS in 57m 58s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cadddb328fa44f46b0d4adc4cfe113e4 : SUCCESS in 36m 08s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0730c5a836a54b229f51bd5abaf34a18 : SUCCESS in 29m 05s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb0fb5f1eaa648e2812b2d7961ff6e2d : SUCCESS in 1h 17m 09s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/38394fa2679642089f3a100ad163e14a : SUCCESS in 1h 25m 45s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b6981ecde4449c4a2b9d1f6a7873e2d : SUCCESS in 17m 06s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/feaa4f4fd1a14aef9566aa22434d17bb : SUCCESS in 11m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a261ec43f3e43a3b67cab32c7e902be : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4c39a3a605bfac50b72a4cb717a984a4a28e693c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 09:35:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3ffb64bc39caffcd8899bba5b645c018aa9a67af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 11:10:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a04fa6f74d09425ea1e812a1243d56ea\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa788f8e91194e0cba5cc49f55dbefd9 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db132dd1de7e4e4bb400b10282ffe79e : SUCCESS in 2m 43s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/824e32ad569e493b8b25a5dccd1ae7e4 : FAILURE in 8m 53s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c9c1769ec3f4193867535dbb78190b2 : FAILURE in 5m 32s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94ea9996c6354fa3887417a3fb336e94 : FAILURE in 5m 28s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fa5076f6bb0426eaca96f6f864de7b3 : SUCCESS in 28m 50s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b95aa33b965e4a47bd74f34002b187af : SUCCESS in 1h 10m 12s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e420374780644b3ba411239684e3552c : SUCCESS in 51m 11s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11492f080b2d43efaec9677d689cea66 : SUCCESS in 30m 02s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97773d33bae84e8ab738621694e3c20f : SUCCESS in 1h 22m 03s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af848b16e4654e5cb90bd7bca86d59fd : SUCCESS in 1h 29m 15s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef238f33339d436da3fccee35abfcf51 : SUCCESS in 16m 58s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd011ebcda2b4e78b98b939ae2b924ac : SUCCESS in 13m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c9b912865394e00beae866c300491d1 : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1dfcded2f0df8f8593bbc6b0929715649a9f5432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 12:43:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d83dfb8d4e7654e89a57628ab8428f01cc0b6b07","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-21 12:48:27.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"d06cdbaa842d076159148b417d85af078e8e15a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 14:09:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb3488c64694f8f89acad9561fdc44c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0ffc64e6ce241289faf63f3434f3fab : SUCCESS in 3m 44s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8d3247d98e541108f2c3d5bae831ed8 : SUCCESS in 3m 03s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/39a4b58d5def44f195d2a162d354095b : SUCCESS in 4m 33s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/912e15e20620478faba1afe9e6bed9ad : SUCCESS in 5m 27s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7104d7310c11465d841dbd565444c061 : SUCCESS in 5m 14s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a82ef70495d446e2936d01504ecf762f : SUCCESS in 43m 43s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca72695171ec45eeb5aa0988ca70b3fe : SUCCESS in 59m 19s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4c62c2ca491433592eabf296f1b2ae7 : SUCCESS in 52m 51s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90f713b3728c465ba5f5f7ecb75efc48 : SUCCESS in 1h 19m 19s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/918ede8cbb1d4c21bd3e80a822b0126a : SUCCESS in 45m 18s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a343aa913a234f70ab912415a717345a : SUCCESS in 44m 56s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/091287dad82f4887973fa13bb034865e : SUCCESS in 10m 04s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52f0f98ff87a46f097551db572eba851 : SUCCESS in 10m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55c9910e5bf74825815d0b02ac4a8a0c : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2cb5e6b1ec8f9dabc29bd21ff7c12d5513645f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-03 14:40:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"08d10e63e99eaf2f3e40081457bf451f4642f1a7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-03 14:58:45.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"2369f8c71c61df58edc16f214e8efc72ed9ddbb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:12:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1c7d9c5a684154ac444b17a1b42d19\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3eb65e387f64692bcbb23fb33635133 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5620d5a22c4d48c082ff79fcaba80cb6 : SUCCESS in 2m 27s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/44e6c45a030c4b1fa9910c432a5d225e : SUCCESS in 6m 17s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8d755458a0a40129b681fe12cb91fd9 : SUCCESS in 5m 34s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a161d7a470104b9a8dd93dfbaa613180 : SUCCESS in 4m 19s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9bdf6ec2b2643a29a11e55f61ffe474 : SUCCESS in 1h 01m 20s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36e7754dd83047d9b92565ef4c4d8997 : SUCCESS in 1h 10m 12s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/67163ded02254e03b9d606da32f595ae : SUCCESS in 1h 08m 48s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a48f7c31da942fdb8cd8c29d9768f3e : SUCCESS in 1h 05m 54s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48775699a8ec42baa396df10898e3ee5 : SUCCESS in 1h 20m 44s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/92d934d886ab498880a543a38d62ac7b : SUCCESS in 1h 27m 21s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/919cbacd1dac41b6a1409150aa3ebb7b : SUCCESS in 8m 46s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/462bff0e16424a47ad96c01c5fd9e901 : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/747abc0cf6724ac9a648129156038ead : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2c1ce248b075db4153d4796b5a9ff40a0e704186","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 12:58:43.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":9},{"id":"04c246804d451d93bccfbe77a44860e93cb27f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 14:27:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d351f2c4083c42a6893c9b7a8d4c9264\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ddc588b677549e480f03dc7faa81ac7 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31f6db54407a4f37bf682753a5cb41c0 : SUCCESS in 2m 41s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/4994e8c1b5894a6596fe854601bb6578 : SUCCESS in 3m 53s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/662851c8163f4813a8a65e15271723cb : SUCCESS in 4m 13s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5a08b30dfff48cea044684c92c6b7ad : SUCCESS in 5m 44s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a05853a27b394e0abd6a8d85173481c4 : SUCCESS in 58m 26s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c12b59a3c93b4a3dbd71a4329e51840a : SUCCESS in 1h 09m 23s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0094a585641a4b8d963696d0019eac42 : SUCCESS in 1h 00m 07s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d49da2dbe4b4eca93dfe4bae88b8030 : SUCCESS in 38m 19s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86c997448b1a4e8dbaeb253ad84f10bd : SUCCESS in 1h 17m 52s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f0c038e07d644e18e8e16c09deec611 : SUCCESS in 1h 23m 29s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1daf5cd429b740f083a3b54dd2a8e2a4 : SUCCESS in 16m 32s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/891ce831b76b49828bf96b5c3df5fa17 : SUCCESS in 11m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efc18abea36d40b0a0cb3f02b5faa44a : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"381d708c5dcba809a9b6d70a4867545eec3a742b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 15:29:07.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"cadaae1ecc3ffc4924c0533c1f325fa5f58ced60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 15:34:35.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"4d58cdcf4edb9173437ccdf038fb55e3efa56d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 16:57:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192eb7bd9e2f4145883337deaab93e07\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6071edf47184f4baff1152afb2a1816 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe1e428aded64f0196203b1f0d09113f : SUCCESS in 3m 00s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/a9f66ec2c8f040d585925332e5210eeb : SUCCESS in 4m 21s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc8653628bee44b399cab1f47858d6cf : SUCCESS in 4m 44s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d5438fa106b44b98a7637b5046e6aba : SUCCESS in 6m 24s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cda6ce2f6fc14120aa01c0daf7e9dd45 : SUCCESS in 53m 14s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/549da7e24e304f3cbababd561a9e487d : SUCCESS in 1h 02m 46s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7beaf31308634a50a1fc78262a41a3bb : SUCCESS in 1h 07m 07s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f76b17b6e6b49559474a5f1cf76d169 : SUCCESS in 32m 35s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/13d15f08c5ec415dbe72ad6f2896e904 : SUCCESS in 1h 06m 37s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dff34c8594ec42f498e6268bb978f128 : SUCCESS in 1h 17m 12s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3830f4222abb431f985395ff7de083e0 : SUCCESS in 17m 07s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d0a8b4dcf974fb29b39e3143a6e2230 : SUCCESS in 11m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f862bec715c7485f96215f9fbdc11f0a : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9ae5731e1f69b7df7b8aa30925a3129c01f9904a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-27 05:51:03.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"b3152ce170bc1c3e007b0960689279eea3ea4fc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 07:13:09.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbb02e207a724f3dbd26ffe8cba7ae48\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cedf934984c4fd2a35d15d716a25952 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccc44bdf2ba34604875f1b0c1468d0b6 : SUCCESS in 2m 35s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/18e28fe526d94bd9aa5bb68ea8ebb18e : SUCCESS in 3m 52s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31b55044e95e48e7904225f2c9311a49 : SUCCESS in 4m 16s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c946c6a46104a9ba5698e27614129c1 : SUCCESS in 4m 47s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e591b98b7bf4da48d881230fb113d68 : SUCCESS in 45m 39s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6318f63ec6fb4e898b4c2ae5cb8e7f2d : SUCCESS in 54m 01s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a52573fea2e41aba222304c4f3a5051 : SUCCESS in 1h 03m 43s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1832167357b24a80a7335da7e6926c27 : SUCCESS in 1h 01m 49s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8dbd0c2dfdfd4ae5987dc6176d124419 : SUCCESS in 1h 06m 27s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28fdb1d4f5c94a47a9a09e6173d4f664 : SUCCESS in 1h 17m 36s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c992182fad9f47ec914488ec82896800 : SUCCESS in 14m 38s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca061c58763e47fe83f6eb1096d1d692 : SUCCESS in 11m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb4fc65880344e97b69883dcb3d26e01 : SUCCESS in 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"640e5c13a929c061b7be70c32e222eb9c501c81e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-29 06:18:32.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"d99581188fe4719df4fe597366451277ccab0cba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-09-29 14:17:53.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"1553ffb2df49edfa83a80a8c209d5375c521dee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 14:18:34.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"4a63df1eaff20b0544718f723b1845ff426cc7c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:26:27.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e93dea34f46f476aa0c4f97c1ce41195\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/019c7d499f034dc98bae4273ff225c4d : SUCCESS in 1m 47s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f799790d87d433fb02854b626bf6e25 : SUCCESS in 1m 48s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/275290fdb6664d62a96bde4284caa8e5 : SUCCESS in 4m 17s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c51542a596ad40bab605e4d2fdedf73d : SUCCESS in 6m 10s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48a5b023506f49cc83543176d01d8e0b : SUCCESS in 1h 03m 46s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ea058fd6b9e44a5a766a82368ed2793 : SUCCESS in 47m 27s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cff334584044ca1bec56be65686e68d : SUCCESS in 17m 31s","accounts_in_message":[],"_revision_number":12},{"id":"9c232b347fc0ced6695ecd3ba52fa39bf875c578","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:26:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"c8b7299299ab2017a8df319c1d681a268234c4cb","revisions":{"4ee13d0bbc572b4db476fd27a388887daafbd64d":{"kind":"REWORK","_number":1,"created":"2025-08-21 05:13:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/1"}}},"commit":{"parents":[{"commit":"385e6d60b4f0d1b7c648e93cc218bb0d583d38e8","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/385e6d60b4f0d1b7c648e93cc218bb0d583d38e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:13:11.000000000","tz":540},"subject":"WIP: Refactor api service management","message":"WIP: Refactor api service management\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4ee13d0bbc572b4db476fd27a388887daafbd64d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/4ee13d0bbc572b4db476fd27a388887daafbd64d"}]},"branch":"refs/heads/master"},"38266e1b3606e78e97aad9da7098ebf8a55a454d":{"kind":"REWORK","_number":2,"created":"2025-08-21 05:25:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/2"}}},"commit":{"parents":[{"commit":"385e6d60b4f0d1b7c648e93cc218bb0d583d38e8","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/385e6d60b4f0d1b7c648e93cc218bb0d583d38e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:25:47.000000000","tz":540},"subject":"WIP: Refactor api service management","message":"WIP: Refactor api service management\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/38266e1b3606e78e97aad9da7098ebf8a55a454d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/38266e1b3606e78e97aad9da7098ebf8a55a454d"}]},"branch":"refs/heads/master"},"d12f264c6163b784486a2e239708caa4db17ae23":{"kind":"REWORK","_number":3,"created":"2025-08-21 05:45:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/3"}}},"commit":{"parents":[{"commit":"385e6d60b4f0d1b7c648e93cc218bb0d583d38e8","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/385e6d60b4f0d1b7c648e93cc218bb0d583d38e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:45:21.000000000","tz":540},"subject":"WIP: Refactor api service management","message":"WIP: Refactor api service management\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d12f264c6163b784486a2e239708caa4db17ae23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/d12f264c6163b784486a2e239708caa4db17ae23"}]},"branch":"refs/heads/master"},"5ce70fbd7f523d5673b4bb8f7edb6a6fad5a3958":{"kind":"REWORK","_number":4,"created":"2025-08-21 07:06:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/4"}}},"commit":{"parents":[{"commit":"385e6d60b4f0d1b7c648e93cc218bb0d583d38e8","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/385e6d60b4f0d1b7c648e93cc218bb0d583d38e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 07:06:47.000000000","tz":540},"subject":"WIP: Refactor api service management","message":"WIP: Refactor api service management\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5ce70fbd7f523d5673b4bb8f7edb6a6fad5a3958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/5ce70fbd7f523d5673b4bb8f7edb6a6fad5a3958"}]},"branch":"refs/heads/master"},"7e337a3e0d2ba746a7a00ca14e4ac36fe25be86d":{"kind":"REWORK","_number":5,"created":"2025-08-21 07:15:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/5"}}},"commit":{"parents":[{"commit":"385e6d60b4f0d1b7c648e93cc218bb0d583d38e8","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/385e6d60b4f0d1b7c648e93cc218bb0d583d38e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 07:15:42.000000000","tz":540},"subject":"WIP: Refactor api service management","message":"WIP: Refactor api service management\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7e337a3e0d2ba746a7a00ca14e4ac36fe25be86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/7e337a3e0d2ba746a7a00ca14e4ac36fe25be86d"}]},"branch":"refs/heads/master"},"2287f6bfe985d6f3882558c0c723e049b1e192c3":{"kind":"REWORK","_number":6,"created":"2025-08-21 09:35:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/6"}}},"commit":{"parents":[{"commit":"385e6d60b4f0d1b7c648e93cc218bb0d583d38e8","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/385e6d60b4f0d1b7c648e93cc218bb0d583d38e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 09:35:12.000000000","tz":540},"subject":"WIP: Refactor api service management","message":"WIP: Refactor api service management\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2287f6bfe985d6f3882558c0c723e049b1e192c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/2287f6bfe985d6f3882558c0c723e049b1e192c3"}]},"branch":"refs/heads/master"},"3e5e7248261f2b0da3723189fdf5cb34d644b7e7":{"kind":"REWORK","_number":7,"created":"2025-08-21 12:43:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/7"}}},"commit":{"parents":[{"commit":"385e6d60b4f0d1b7c648e93cc218bb0d583d38e8","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/385e6d60b4f0d1b7c648e93cc218bb0d583d38e8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 12:43:41.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3e5e7248261f2b0da3723189fdf5cb34d644b7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3e5e7248261f2b0da3723189fdf5cb34d644b7e7"}]},"branch":"refs/heads/master"},"c9073d79a19ed695feb6be7823e8c75227946ab0":{"kind":"REWORK","_number":8,"created":"2025-09-03 14:40:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/8"}}},"commit":{"parents":[{"commit":"0734e757a89e0319138f0b67e23aef4a319bb559","subject":"Merge \"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/0734e757a89e0319138f0b67e23aef4a319bb559"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-03 14:39:32.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c9073d79a19ed695feb6be7823e8c75227946ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c9073d79a19ed695feb6be7823e8c75227946ab0"}]},"branch":"refs/heads/master"},"cfcc8dd850342d1800550e395e19a1000be59303":{"kind":"REWORK","_number":9,"created":"2025-09-26 12:58:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/9"}}},"commit":{"parents":[{"commit":"07ff154c4dd4fff271019977a2c64c727784c0e3","subject":"Follow up \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/07ff154c4dd4fff271019977a2c64c727784c0e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 12:58:21.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/cfcc8dd850342d1800550e395e19a1000be59303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/cfcc8dd850342d1800550e395e19a1000be59303"}]},"branch":"refs/heads/master"},"30bfb0e122d69cd6131827cb532c8ff1fb2c02eb":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2025-09-26 15:29:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/10"}}},"commit":{"parents":[{"commit":"07ff154c4dd4fff271019977a2c64c727784c0e3","subject":"Follow up \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/07ff154c4dd4fff271019977a2c64c727784c0e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 15:29:07.000000000","tz":0},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/30bfb0e122d69cd6131827cb532c8ff1fb2c02eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/30bfb0e122d69cd6131827cb532c8ff1fb2c02eb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3b3169492d1fa7391b867427961c13ea96603d23":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2025-09-26 15:34:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/11"}}},"commit":{"parents":[{"commit":"07ff154c4dd4fff271019977a2c64c727784c0e3","subject":"Follow up \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/07ff154c4dd4fff271019977a2c64c727784c0e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 15:34:35.000000000","tz":0},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3b3169492d1fa7391b867427961c13ea96603d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/3b3169492d1fa7391b867427961c13ea96603d23"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c8b7299299ab2017a8df319c1d681a268234c4cb":{"kind":"REWORK","_number":12,"created":"2025-09-27 05:51:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/958140/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-aodh","ref":"refs/changes/40/958140/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-aodh refs/changes/40/958140/12"}}},"commit":{"parents":[{"commit":"07ff154c4dd4fff271019977a2c64c727784c0e3","subject":"Follow up \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/07ff154c4dd4fff271019977a2c64c727784c0e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-21 05:12:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-27 05:50:48.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I179246eabd540e7c905d67eb3f99840809c99e4e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c8b7299299ab2017a8df319c1d681a268234c4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-aodh/commit/c8b7299299ab2017a8df319c1d681a268234c4cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
