)]}'
{"id":"openstack%2Fpuppet-barbican~659997","triplet_id":"openstack%2Fpuppet-barbican~master~I0484413db8202fa1edcb87f525684f679131d1a3","project":"openstack/puppet-barbican","branch":"master","topic":"fix-client-test","hashtags":[],"change_id":"I0484413db8202fa1edcb87f525684f679131d1a3","subject":"Fix client spec test","status":"MERGED","created":"2019-05-20 03:03:52.000000000","updated":"2019-05-20 04:38:08.000000000","submitted":"2019-05-20 04:38:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659997-1558327088559-334d6ac5","meta_rev_id":"7fce48bf06190d819f7a95e7a5a1c2cb9cd9eb8d","_number":659997,"virtual_id_number":659997,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-20 04:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-20 04:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 04:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 04:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82bc6de028536272cf9744c1b59fe900ef6b01eb","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 03:03:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"485c92f70a5bfd4dd81112b417cc606359b21025","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:17:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/97/659997/1/check/puppet-openstack-lint-ubuntu-bionic/2534709/ : SUCCESS in 3m 44s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/97/659997/1/check/puppet-openstack-syntax-4-ubuntu-bionic/b68081b/ : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/97/659997/1/check/puppet-openstack-syntax-5-ubuntu-bionic/c6f4b21/ : SUCCESS in 6m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/97/659997/1/check/puppet-openstack-syntax-6-ubuntu-bionic/f415616/ : SUCCESS in 4m 38s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/97/659997/1/check/puppet-openstack-unit-4.8-centos-7/0c6bd6a/ : SUCCESS in 11m 07s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/97/659997/1/check/puppet-openstack-unit-5.5-centos-7/dc8447a/ : SUCCESS in 8m 55s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/97/659997/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/0991de9/ : SUCCESS in 10m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/97/659997/1/check/puppet-openstack-unit-latest-ubuntu-bionic/aab975b/ : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d453ef418ef9b93917cb8e8d55a863a0fde959a7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 04:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2dae7fe547b17e389bf0028e24c9076f448fb0e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:18:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7db92c4160e201ff0a616c073e1ce45f64760b55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:38:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d9d503272fec0bfc0ee1f42419f6a54ed1d4e5dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:38:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/97/659997/1/gate/puppet-openstack-lint-ubuntu-bionic/3ebad38/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/97/659997/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/b7795cc/ : SUCCESS in 3m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/97/659997/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/37f4b5b/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/97/659997/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/fa04a04/ : SUCCESS in 3m 51s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/97/659997/1/gate/puppet-openstack-unit-4.8-centos-7/5a6d3b9/ : SUCCESS in 19m 00s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/97/659997/1/gate/puppet-openstack-unit-5.5-centos-7/3030417/ : SUCCESS in 17m 59s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/97/659997/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/a7456f3/ : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae7dbe45265f7c6bcff3bb3d562620c2104e442e","revisions":{"ae7dbe45265f7c6bcff3bb3d562620c2104e442e":{"kind":"REWORK","_number":1,"created":"2019-05-20 03:03:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/97/659997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/97/659997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/97/659997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/97/659997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/97/659997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/97/659997/1"}}},"commit":{"parents":[{"commit":"2e25f8f734502385a3214c04736e393f6c455019","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/2e25f8f734502385a3214c04736e393f6c455019"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:18.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:18.000000000","tz":480},"subject":"Fix client spec test","message":"Fix client spec test\n\nChange-Id: I0484413db8202fa1edcb87f525684f679131d1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ae7dbe45265f7c6bcff3bb3d562620c2104e442e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ae7dbe45265f7c6bcff3bb3d562620c2104e442e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
