)]}'
{"id":"openstack%2Fpuppet-barbican~669035","triplet_id":"openstack%2Fpuppet-barbican~master~I2beb72f1ab37130eca340e691ca2dfd15cb5aa61","project":"openstack/puppet-barbican","branch":"master","topic":"bug/1835161","hashtags":[],"change_id":"I2beb72f1ab37130eca340e691ca2dfd15cb5aa61","subject":"Manage oslo_middleware/max_request_body_size","status":"MERGED","created":"2019-07-04 03:44:21.000000000","updated":"2019-07-05 01:52:52.000000000","submitted":"2019-07-05 01:52:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669035-1562291572806-769ffea3","meta_rev_id":"22c49cd82225bfc2ac2f37d42f3c6799c5c9fa2d","_number":669035,"virtual_id_number":669035,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-05 01:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-04 13:59:22.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-05 00:51:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 13:59:22.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-05 01:52:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f26d845f56d34451266ead291688bda8b032a567","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 03:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93e2c28238bff16b8e484aa16776bca2065f1cc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 04:51:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/35/669035/1/check/puppet-openstack-lint-ubuntu-bionic/20c430c/ : SUCCESS in 6m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/35/669035/1/check/puppet-openstack-syntax-5-ubuntu-bionic/355f20e/ : SUCCESS in 7m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/35/669035/1/check/puppet-openstack-syntax-6-ubuntu-bionic/183394f/ : SUCCESS in 6m 53s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/35/669035/1/check/puppet-openstack-unit-5.5-centos-7/74818d2/ : FAILURE in 23m 48s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/35/669035/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/6e89149/ : FAILURE in 13m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/35/669035/1/check/puppet-openstack-unit-latest-ubuntu-bionic/a9f07fa/ : FAILURE in 10m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/35/669035/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/a84cd8f/ : FAILURE in 6m 37s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/35/669035/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/28721a9/ : SUCCESS in 59m 59s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/35/669035/1/check/puppet-openstack-beaker-centos-7/7b4a5cb/ : SUCCESS in 17m 47s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/35/669035/1/check/puppet-openstack-beaker-ubuntu-bionic/4bad44e/ : FAILURE in 8m 32s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/35/669035/1/check/build-openstack-releasenotes/8975a06/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6465cc92e06378079c797bab9cf7f998f0eef719","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 05:34:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9740c0ec02b63eff01183e254a747819cdd5a569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 06:30:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/35/669035/2/check/puppet-openstack-lint-ubuntu-bionic/dab84ed/ : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/35/669035/2/check/puppet-openstack-syntax-5-ubuntu-bionic/80a164a/ : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/35/669035/2/check/puppet-openstack-syntax-6-ubuntu-bionic/1a8bc6e/ : SUCCESS in 6m 48s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/35/669035/2/check/puppet-openstack-unit-5.5-centos-7/64f30e0/ : SUCCESS in 16m 33s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/35/669035/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/7694f6c/ : SUCCESS in 11m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/35/669035/2/check/puppet-openstack-unit-latest-ubuntu-bionic/384e35b/ : SUCCESS in 7m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/35/669035/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/570266d/ : FAILURE in 3m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/35/669035/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/6b84ff0/ : SUCCESS in 48m 58s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/35/669035/2/check/puppet-openstack-beaker-centos-7/1b65ad9/ : SUCCESS in 16m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/35/669035/2/check/puppet-openstack-beaker-ubuntu-bionic/482dbb2/ : FAILURE in 6m 49s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/35/669035/2/check/build-openstack-releasenotes/006bf24/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"3dd4470648bd5217af3d1709cd037d762e62598f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 13:59:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"67269b18b3299f08ecf8a24d4f3d1507fc143fb6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-05 00:51:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"808640fce9aec1bda65bb06eae41ee8acd2fc023","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 00:51:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ea85ea895f15ae7444421f4dfa24654c8345d85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 01:52:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/35/669035/2/gate/puppet-openstack-lint-ubuntu-bionic/63a970e/ : SUCCESS in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/35/669035/2/gate/puppet-openstack-syntax-5-ubuntu-bionic/db15a82/ : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/35/669035/2/gate/puppet-openstack-syntax-6-ubuntu-bionic/9db7a06/ : SUCCESS in 4m 31s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/35/669035/2/gate/puppet-openstack-unit-5.5-centos-7/90f7835/ : SUCCESS in 10m 59s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/35/669035/2/gate/puppet-openstack-unit-6.4-ubuntu-bionic/d6720f6/ : SUCCESS in 10m 24s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/35/669035/2/gate/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/790c052/ : FAILURE in 4m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/35/669035/2/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/5b68432/ : SUCCESS in 54m 59s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/35/669035/2/gate/puppet-openstack-beaker-centos-7/53926c2/ : SUCCESS in 16m 35s\n- build-openstack-releasenotes http://logs.openstack.org/35/669035/2/gate/build-openstack-releasenotes/6fab3ee/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1b2f56676d60282d9c2edd3fb1fe0a6768765b0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 01:52:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"def436a021fe877d498d9152e2893d9a5d493865","revisions":{"09d63b3c4fede49f54f1c748c6986c85d245f5a3":{"kind":"REWORK","_number":1,"created":"2019-07-04 03:44:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/35/669035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/35/669035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/1"}}},"commit":{"parents":[{"commit":"0645cd442130a5a65283709c8cfde8132b758a45","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0645cd442130a5a65283709c8cfde8132b758a45"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:34:41.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:35:54.000000000","tz":480},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached barbican will return a 413 Entity Too Large\nand not log anything to barbican.log.\n\nChange-Id: I2beb72f1ab37130eca340e691ca2dfd15cb5aa61\nCloses-Bug: #1835161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/09d63b3c4fede49f54f1c748c6986c85d245f5a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/09d63b3c4fede49f54f1c748c6986c85d245f5a3"}]},"branch":"refs/heads/master"},"def436a021fe877d498d9152e2893d9a5d493865":{"kind":"REWORK","_number":2,"created":"2019-07-04 05:34:17.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/35/669035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/35/669035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/35/669035/2"}}},"commit":{"parents":[{"commit":"0645cd442130a5a65283709c8cfde8132b758a45","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0645cd442130a5a65283709c8cfde8132b758a45"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:34:41.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 05:34:07.000000000","tz":480},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached barbican will return a 413 Entity Too Large\nand not log anything to barbican.log.\n\nChange-Id: I2beb72f1ab37130eca340e691ca2dfd15cb5aa61\nCloses-Bug: #1835161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/def436a021fe877d498d9152e2893d9a5d493865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/def436a021fe877d498d9152e2893d9a5d493865"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
