)]}'
{"id":"openstack%2Fpuppet-barbican~677631","triplet_id":"openstack%2Fpuppet-barbican~master~I3debab140115a91f3df7aabf00c87eb1842b293b","project":"openstack/puppet-barbican","branch":"master","topic":"bug/1840868","hashtags":[],"change_id":"I3debab140115a91f3df7aabf00c87eb1842b293b","subject":"Introduce the new rabbit_heartbeat_in_pthread option","status":"MERGED","created":"2019-08-21 06:25:19.000000000","updated":"2019-08-22 03:28:34.000000000","submitted":"2019-08-22 03:28:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677631-1566444514656-848e10b7","meta_rev_id":"9f7a31a865a3ec24a816f451e91be38a4652ebf7","_number":677631,"virtual_id_number":677631,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-22 03:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 02:23:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-21 15:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-22 02:23:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 15:16:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-22 02:23:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 03:28:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"748bf25a1c4b3cff3998a4650cd4772aa49fe6a8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-21 06:25:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1319c3e5c93a99905f5f17f712f9351087c87feb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:41:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2229ac1ca1de47c7b7fda134f11c97a8 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db90d122dc414bbeb9e64748337a67b8 : SUCCESS in 5m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c59101221da34670b0a1375ab6859666 : SUCCESS in 7m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/336977c596094c3b9b0e0038963ba8a4 : SUCCESS in 19m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/522c2fa2ccfe49ff926cbe13f834d955 : SUCCESS in 14m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0908d01bd8ab47e7a97e4a98f9867d40 : SUCCESS in 12m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50694781e077434e9884c95a5a1ee277 : FAILURE in 31m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e463705379934f9d92fd91da4e16eadb : SUCCESS in 1h 03m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f9fca61c8d4b4cf096af77f02fff36b2 : SUCCESS in 16m 45s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd37a83239e2466e8df18662bad62499 : FAILURE in 10m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/415533d7e4274eaf86b412994f638858 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4c34b16ce7c395e2e4cbd9ab7192284328abdaf4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-21 15:16:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6b317fec70392547ee5ba2b32fd27251ded2cc25","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:23:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ed0be336e268c36275bc78c8cb9f3ff05fa8bfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 02:23:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37dc0d3e91c2ece2ba8120ad5462e9f060a86868","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:28:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c97c9f5cc104af42c4ccbc50afcd87eb187263e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:28:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/946cf807a79b495abe7fb9cbed91dedd : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e304e76b61de479e9ad6b20e07bfc444 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef27263db8a4429087f48ce79217be86 : SUCCESS in 5m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e7cfa91a535f413ea7b1058cdfa9e3c9 : SUCCESS in 19m 19s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4562e7afaf0484990eafa91348769e9 : SUCCESS in 14m 40s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c86ef178c36140a0ab677acea2cb81b9 : FAILURE in 34m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ffb6248c2b8a49f4b35fabe8034da5e4 : SUCCESS in 58m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/525493622d6f487d9bf4ec6b9bebd631 : SUCCESS in 17m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/071331fa771744de950aad192e271e03 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3471b1b931dc9a1ef20bacc7cff808334e34228","revisions":{"b3471b1b931dc9a1ef20bacc7cff808334e34228":{"kind":"REWORK","_number":1,"created":"2019-08-21 06:25:19.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/31/677631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/31/677631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/31/677631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/31/677631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/31/677631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/31/677631/1"}}},"commit":{"parents":[{"commit":"39fa246cd00699a44216bb59f23fa347cd9e0164","subject":"Merge \"Configure keystone_authtoken/service_token_roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/39fa246cd00699a44216bb59f23fa347cd9e0164"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:21:00.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:22:41.000000000","tz":480},"subject":"Introduce the new rabbit_heartbeat_in_pthread option","message":"Introduce the new rabbit_heartbeat_in_pthread option\n\noslo.messaging RabbitMQ driver have now a new option that allow user to\nrun the RabbitMQ heartbeat over a native python thread.\n\nThese change allow user to use this new option.\n\nChange-Id: I3debab140115a91f3df7aabf00c87eb1842b293b\nCloses-Bug: #1840868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/b3471b1b931dc9a1ef20bacc7cff808334e34228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/b3471b1b931dc9a1ef20bacc7cff808334e34228"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
