)]}'
{"id":"openstack%2Fpuppet-barbican~688014","triplet_id":"openstack%2Fpuppet-barbican~master~I985341ca95807763b4ddd6653dbfcfdc87fb903a","project":"openstack/puppet-barbican","branch":"master","hashtags":[],"change_id":"I985341ca95807763b4ddd6653dbfcfdc87fb903a","subject":"Update the constraints url","status":"MERGED","created":"2019-10-11 00:54:50.000000000","updated":"2019-10-16 14:05:31.000000000","submitted":"2019-10-16 14:03:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688014-1571234608574-d25b5e01","meta_rev_id":"cff382b96a73d493054c6bc9143af0a77af5956d","_number":688014,"virtual_id_number":688014,"owner":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 14:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-10-16 13:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2019-10-16 13:02:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 13:02:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-16 14:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0dee601dd10c495fca44203794f2405fa77863a","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-10-11 00:54:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e9dcf11746e154c127395ee22a380f59e228937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 02:09:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7ba845a8d7d45a8bcf7cb9fd2063f8d : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a5577b7237c448b971afcda8b39f6c4 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7cd768c216d44841aabe2b98e124497d : SUCCESS in 5m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b2c826a6e62043249c0ee43d22a9aee6 : SUCCESS in 15m 55s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e979b9c2105472eb76ef082b89949fe : SUCCESS in 11m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f1de1930e8d4eaa8e26a06661b75a9c : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ad9a276a77d46abb622acbffed3e2fe : FAILURE in 37m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/01985675cd7244d886203433d144d327 : FAILURE in 1h 11m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9b3ee3f3196c4033bae4e62ea66b77b7 : SUCCESS in 21m 10s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d002c0b1be754dd599b306c9eaa4c1bb : FAILURE in 11m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a1e0ac59bc142fc883cf036f5fedd03 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"957b036b33c8d4e21c79f1dc07611ca06ebe3247","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-16 11:49:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9eed4a068846f11fde5203a9133c3bdfaaa0a386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 13:00:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52bc77596c8f4e21b079fdb59eebd8a1 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30c48f61361647f89aaede55dd48a660 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca9015db6fca40648b92d1df6c76f0e0 : SUCCESS in 3m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/45fef5d132a94b328e4d78ea327d3f19 : SUCCESS in 17m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a55055bfd0524076ae80a3075fee51fd : SUCCESS in 13m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38cb2146fc9048f3bddad64b9549d9b9 : SUCCESS in 10m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f2d4c1366df43af97de046c92f0fa5f : FAILURE in 30m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9867c4935ade4dce9ffe05ed2decf1e5 : SUCCESS in 1h 05m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1d4e800bcaa2404598ab9cf7b329cd51 : SUCCESS in 15m 39s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c742122f08744129a3907ba873caa9f : FAILURE in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/907c5245269649758770ba4f18ca8052 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0b8e45b2ce079a3b0726aaca8407daffbc3aa340","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-16 13:02:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbaa6f71c513ddc6eea076b5e2d5d7479e9add7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 13:02:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2cbd1314a99def552a5c9a38b3800e59b6ed12b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:03:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/895ce304e8a8483fbfadddc4bc788bfa : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26690486562e40db89db11e2bf33d680 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db38e900ee0c4994af6bba57da7d5c87 : SUCCESS in 4m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4baa1a4942c5490cb0b3d9ea222e345e : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56793d66793d48528b5df494132446d4 : SUCCESS in 12m 09s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04a747bfcf5a47c1b5e986b994fc4014 : FAILURE in 28m 20s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/35bdfb882e19416fb2c2649c8cf0a15b : SUCCESS in 55m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d49307cefdd34d418477b0173a4bee2d : SUCCESS in 17m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0221cb1bb42e4b0cbc04192b61af0616 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0c22f1e716a22e1a2a64a9a736b6145a76a6c538","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:03:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cff382b96a73d493054c6bc9143af0a77af5956d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:05:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79e4240205f34141b967363c414ae13c : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4b07dd193d125967bfdfee5cf7177bc98916ef5","revisions":{"a4b07dd193d125967bfdfee5cf7177bc98916ef5":{"kind":"REWORK","_number":1,"created":"2019-10-11 00:54:50.000000000","uploader":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"ref":"refs/changes/14/688014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/14/688014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/14/688014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/14/688014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/14/688014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/14/688014/1"}}},"commit":{"parents":[{"commit":"c21311f9150b33e2eb825bb702b8a1201221c104","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/c21311f9150b33e2eb825bb702b8a1201221c104"}]}],"author":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-11 00:53:27.000000000","tz":480},"committer":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-11 00:54:46.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nChange-Id: I985341ca95807763b4ddd6653dbfcfdc87fb903a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a4b07dd193d125967bfdfee5cf7177bc98916ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a4b07dd193d125967bfdfee5cf7177bc98916ef5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
