)]}'
{"id":"openstack%2Fpuppet-barbican~689287","triplet_id":"openstack%2Fpuppet-barbican~master~I4cb1e130887fb58d57cee455ee0ef72a7a713e94","project":"openstack/puppet-barbican","branch":"master","hashtags":[],"change_id":"I4cb1e130887fb58d57cee455ee0ef72a7a713e94","subject":"add bug link to readme","status":"ABANDONED","created":"2019-10-18 01:15:32.000000000","updated":"2019-10-18 08:35:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fbe4f13fc61180b7ba1dab706ff66927a384a568","_number":689287,"virtual_id_number":689287,"owner":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-18 08:15:52.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-18 01:24:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 01:24:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-18 08:15:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"64ebfca93508a6b18464a9c4df7d2867dc48cac7","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-10-18 01:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45efbcfbad21227927f292380998295f5a2aa312","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-10-18 01:16:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c88a1764f1c5567fff9a071b2954608b83f35286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 01:24:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d7899469f5244fe80ca8687afbe82d2 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7823303beaa7691e96f90fb6edd3d17a31575884","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-18 08:15:52.000000000","message":"Patch Set 2:\n\nwrong url, also we are hoping to move bugs to storyboard in the future so would need to redo this anyways if the url was correct","accounts_in_message":[],"_revision_number":2},{"id":"fbe4f13fc61180b7ba1dab706ff66927a384a568","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-10-18 08:35:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"65135001e37b872e87ab9727ba323ca7c5c0bc27","revisions":{"95750094ec9db909e52256a4aaededb24717c8d0":{"kind":"REWORK","_number":1,"created":"2019-10-18 01:15:32.000000000","uploader":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"ref":"refs/changes/87/689287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/87/689287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/1"}}},"commit":{"parents":[{"commit":"a4b07dd193d125967bfdfee5cf7177bc98916ef5","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a4b07dd193d125967bfdfee5cf7177bc98916ef5"}]}],"author":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-18 01:15:12.000000000","tz":480},"committer":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-18 01:15:12.000000000","tz":480},"subject":"add bug link to readmewq","message":"add bug link to readmewq\n\nChange-Id: I4cb1e130887fb58d57cee455ee0ef72a7a713e94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/95750094ec9db909e52256a4aaededb24717c8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/95750094ec9db909e52256a4aaededb24717c8d0"}]},"branch":"refs/heads/master"},"65135001e37b872e87ab9727ba323ca7c5c0bc27":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-18 01:16:06.000000000","uploader":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"ref":"refs/changes/87/689287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/87/689287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/87/689287/2"}}},"commit":{"parents":[{"commit":"a4b07dd193d125967bfdfee5cf7177bc98916ef5","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a4b07dd193d125967bfdfee5cf7177bc98916ef5"}]}],"author":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-18 01:15:12.000000000","tz":480},"committer":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-18 01:15:45.000000000","tz":480},"subject":"add bug link to readme","message":"add bug link to readme\n\nChange-Id: I4cb1e130887fb58d57cee455ee0ef72a7a713e94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/65135001e37b872e87ab9727ba323ca7c5c0bc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/65135001e37b872e87ab9727ba323ca7c5c0bc27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
