)]}'
{"id":"openstack%2Fpuppet-barbican~715787","triplet_id":"openstack%2Fpuppet-barbican~master~I687129bbc5a2ab344db1fcb29c244412d29fa375","project":"openstack/puppet-barbican","branch":"master","hashtags":[],"change_id":"I687129bbc5a2ab344db1fcb29c244412d29fa375","subject":"Add CentOS 8 and Puppet 6 in metadata","status":"MERGED","created":"2020-03-30 07:17:13.000000000","updated":"2020-03-31 14:36:33.000000000","submitted":"2020-03-31 14:36:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715787-1585665393432-93ed6149","meta_rev_id":"8a87c4b08b24c7352b12bef24456a6c01915b2ee","_number":715787,"virtual_id_number":715787,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 14:36:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-30 13:52:43.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-31 14:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-31 14:29:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:52:43.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-31 14:29:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-31 14:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7517ca476ec91b84ebdc8c10f7684875797d0c54","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07a04da9416ce8ed40fad9b37a402b455e586ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 07:24:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/97a1416c2cd04e4ea85be62ff8695d27 : SUCCESS in 2m 19s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8f9d655585c4025bbd44e1e1dd8eba3 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"537c73d838eb132b090b83d19ea9a6b126f9904e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:42:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7fe11cd186d8d453c966d5e9415ba3b178367eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 07:53:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/695c6f055a5f4047ab372c1b4bda160a : SUCCESS in 4m 21s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a22a933bdde4f1dba4efc88caae5dd3 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d249bd54e39bf95f8cd4cd2b6c81173ee1f133eb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-30 13:52:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9f332747093b59e88cc6a13827db614df7e7cb61","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-31 14:29:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"be6d428e753c151ede57be351c9820579931b4cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:29:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1fbe3e0f0bbdcce2c225bfad0168b4f413e338c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:36:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/20c1de4596af4ff285ce56b3b0847eee : SUCCESS in 3m 22s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8adf61ccc7844edab4811da25605124e : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"d2835d8e0699030fd9409e2fed83f7b055bc05ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:36:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f1a09e50a55c1e5770038a65dfdd7fbe546fac71","revisions":{"4ce97bd0136c3c82cbe688e89bf3df4c6c5d1025":{"kind":"REWORK","_number":1,"created":"2020-03-30 07:17:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/87/715787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/87/715787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/1"}}},"commit":{"parents":[{"commit":"a82d3e4572b5744fdfbefc68e02e268286edc3d9","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a82d3e4572b5744fdfbefc68e02e268286edc3d9"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:17:04.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:17:04.000000000","tz":120},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: I687129bbc5a2ab344db1fcb29c244412d29fa375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4ce97bd0136c3c82cbe688e89bf3df4c6c5d1025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4ce97bd0136c3c82cbe688e89bf3df4c6c5d1025"}]},"branch":"refs/heads/master"},"f1a09e50a55c1e5770038a65dfdd7fbe546fac71":{"kind":"REWORK","_number":2,"created":"2020-03-30 07:42:59.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/87/715787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/87/715787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/87/715787/2"}}},"commit":{"parents":[{"commit":"a82d3e4572b5744fdfbefc68e02e268286edc3d9","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a82d3e4572b5744fdfbefc68e02e268286edc3d9"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:17:04.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:42:56.000000000","tz":0},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: I687129bbc5a2ab344db1fcb29c244412d29fa375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/f1a09e50a55c1e5770038a65dfdd7fbe546fac71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/f1a09e50a55c1e5770038a65dfdd7fbe546fac71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
