)]}'
{"id":"openstack%2Fpuppet-barbican~723931","triplet_id":"openstack%2Fpuppet-barbican~stable%2Fussuri~Ia9d1ea970d82e115b3ffeb32594682de165b4aec","project":"openstack/puppet-barbican","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ia9d1ea970d82e115b3ffeb32594682de165b4aec","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-28 14:49:12.000000000","updated":"2020-04-29 02:11:15.000000000","submitted":"2020-04-29 02:09:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723931-1588126199048-0ef6f594","meta_rev_id":"0d6902bfc150b7df96ced6e8be7ef3fc2cbc5e57","_number":723931,"virtual_id_number":723931,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 02:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 00:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-28 16:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 00:51:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:44:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 00:51:56.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-29 02:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1b1c64894dcf023b695d10a9e76b9ede3d9bb95","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:49:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4be09e36d323ab998c2ae0dc80e767924ed4d6c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:44:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e59a689f59cf3ca1819dbd4466415cdef90fdbf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:57:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/843141ddf13542e0a39258f6fd17b1e1 : SUCCESS in 6m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7753413f01742b1bb882ff72bbd0f32 : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2901677ac6a43889b2aba732a64f923 : SUCCESS in 8m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ce6f902533c54d2f92fe8c2c9ef56e77 : SUCCESS in 16m 07s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/ae7519e1cac94e4ba044ed658e378378 : SUCCESS in 16m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6154bc0481664997901cd5eb8c1a5378 : SUCCESS in 11m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7043b5be7da042c5b88b4dbcc2c903ad : SUCCESS in 13m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52b1cdd810804da99a0f9f79a3bcf3f8 : SUCCESS in 59m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0ac4043f68c04c5b8c0e517898f912ab : SUCCESS in 1h 07m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a970d4d158b14ac69b58ed0589c29d71 : SUCCESS in 23m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ed256b068d944463b6497f61f9de6119 : FAILURE in 12m 14s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e620d6beafa446dc92414e48fdd84a7e : FAILURE in 11m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7945d348f2f041e2ba48b6941a2739f8 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ea0368f1e70a3f3ca8cdaa55e056f6cbf97085a5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 00:51:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82bf0cbc115f7cb492bc9bc5b9ac10bbd2b0b2f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 00:52:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e677f536fdf6182606235c6638fdafdf230a3bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:09:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cf56bd74ded43a9a9de06e6dd0f8a7a : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95e8b8bd8c354e0482aa346d594087b9 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fdc7ff89f3b4bac9bf1d41d093948d2 : SUCCESS in 4m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3c3f1ee622294616824540aa7410f364 : SUCCESS in 10m 40s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d0043aa2293d4ca9b15a05b102a1a586 : SUCCESS in 10m 36s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab7aa61880424ff1b72b99216414f79a : SUCCESS in 8m 23s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb4ea67de41a415f8643db357079f5e1 : SUCCESS in 54m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/58f7b8f701db4aa9bf3945485095d48d : SUCCESS in 1h 05m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8cf2dd9862454da69f3c24dd7a5b3de6 : SUCCESS in 17m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95bdf9129176462d9900be3e8ffad38f : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b30596f03410d63a1e3aad59d4338100e21dcb18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:09:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9daafea55590acc7b784a1676a7b2f085fb2949c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:11:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d5dd15d98a6437b9f04c1e9907a7a2d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8da45e0fa2b45a670f15cdba2a35e1dd28a6556","revisions":{"a8da45e0fa2b45a670f15cdba2a35e1dd28a6556":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:49:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/31/723931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/31/723931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/31/723931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/31/723931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/31/723931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/31/723931/1"}}},"commit":{"parents":[{"commit":"339d0fb2afebf2f4c7e3673baa4c98c82f7b3b42","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/339d0fb2afebf2f4c7e3673baa4c98c82f7b3b42"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:49:10.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:49:10.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ia9d1ea970d82e115b3ffeb32594682de165b4aec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a8da45e0fa2b45a670f15cdba2a35e1dd28a6556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a8da45e0fa2b45a670f15cdba2a35e1dd28a6556"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
