)]}'
{"id":"openstack%2Fpuppet-barbican~731693","triplet_id":"openstack%2Fpuppet-barbican~master~I5de77441cd741b809762166347ec6bef04ba4f4c","project":"openstack/puppet-barbican","branch":"master","hashtags":[],"change_id":"I5de77441cd741b809762166347ec6bef04ba4f4c","subject":"Chain barbican_api_paste_ini to config anchor","status":"MERGED","created":"2020-05-29 08:35:56.000000000","updated":"2020-05-30 07:05:41.000000000","submitted":"2020-05-29 17:48:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731693-1590774516065-3f6755ef","meta_rev_id":"f15b01b0265f398b4ffbf64256343ef02f96d03c","_number":731693,"virtual_id_number":731693,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 17:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-29 11:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-29 13:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2020-05-29 10:24:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-05-29 13:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 10:24:18.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-29 11:42:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-29 13:55:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-29 17:48:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ceb1fa23e78fd43e1d687b291cd88ab1bf370fb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-29 08:35:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a321e814bdfd47fee0fbf7af7f0842bb0c75251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:08:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c729fbc54337418babf95e2c8cae9741 : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3674c78eff644bfa265dc9b0aa84dd2 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93c19df576f440d890268b34dc762fa2 : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/62d119f00fd840cf808fa3746a07c57a : SUCCESS in 15m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b242e7678306478381b813cc8e822a59 : SUCCESS in 12m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdfbd322af7e45629a39ddbf31a581f8 : SUCCESS in 16m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2a3b0218a234291bf89e899e32bf39c : SUCCESS in 9m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa3ffda858cf489fba2590f4082d41bb : SUCCESS in 49m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/adcf637022c441efb9fcfca3f50ab156 : SUCCESS in 1h 31m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f28a505c0dd948c3a9a63f29f327ac52 : SUCCESS in 19m 28s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e5ddaa57abdf429fb520dc6ecad19b83 : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7500295a3f24664badca7d39ad7f0dd : FAILURE in 8m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0664de3f98fe3fc30dee6df0f7a6c7343844322","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-29 11:42:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b5ac3d4a1de9431f665e390954b656e29335924","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-29 13:55:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3671be1eadbfa1db6f36e431f813d09f333e6314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 13:56:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a244665ac67439cf738426446746f6033ae06426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 14:57:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13584cf8ef7144c4a526a2838ecdc762 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/623144c626af4d858b69df66a06414c0 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10b4886b61e14e2094e0b6197eafa765 : SUCCESS in 4m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1e5192de583145dca9b71c7b0736de4c : SUCCESS in 15m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1a29a121025b4570b7f27a79baf4687e : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9eaedd245d0641508db463114a6a878a : SUCCESS in 13m 04s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3642e074675a47f0b18fbf29afa507d0 : SUCCESS in 59m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fbedc6e9b28c4e2c90a7cb2bb196ca64 : FAILURE in 28m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c4611c9c3d1d4dbcb98449f32c3ac150 : SUCCESS in 22m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6a61ac16d39a62d5553502e01eb881f771690aba","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-29 15:11:32.000000000","message":"Patch Set 1:\n\nrecheck\n\nthe error seems to be caused by a temporal issue with delorean repo.","accounts_in_message":[],"_revision_number":1},{"id":"a5fc9a2997797a7808169d9ec489a6243b96c619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 16:18:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4234ff31130242409ffe2c4bd1b965f3 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/339a852239d84379a08528f688003473 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a0cd1899e3948a2be637a074351ff30 : SUCCESS in 9m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ea5dac2f4343427cac53c908d29a660c : SUCCESS in 23m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cee1aef2dbe54b72a007f054a8cf907b : SUCCESS in 17m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/108e088165a4454e91b4216d7a16aefb : SUCCESS in 11m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d56ff489e924079bb66910cd4bdf2bc : SUCCESS in 11m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a808811704f348e899087f462f5cdc4a : SUCCESS in 55m 27s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/de37b6a60c144abcaf66292abdd374a2 : SUCCESS in 1h 05m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e8a43cd6027c4079ac26f3c88ca59831 : SUCCESS in 20m 35s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/bcbe92c2499d4903bfdc94063a5a2047 : FAILURE in 14m 06s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2290483b805c4a71acf1b546e3b7dd5c : FAILURE in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c99907d0a8409bb9ed99cdf16bfefc01081b073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 16:19:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3682dfe20706c4f06b845942e1ffd3e54b1c973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 17:48:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/901fe26ad8a346c4b90ec47647224906 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c26d04b80e06473b994d51e7b42c4d86 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5cbdb0d5db447e7b00ba1e56402d1ba : SUCCESS in 5m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c43ad6bb2cb34e39abad7e8ee11ad878 : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/607b8f75190c43aa8b1888cb095a0145 : SUCCESS in 13m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9675bb2920a94249b8157379e862fc01 : SUCCESS in 13m 51s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/048ea4833dcb422faa07dc578e370615 : SUCCESS in 1h 27m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2493535aecf441498d17ee544d124809 : SUCCESS in 1h 10m 09s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b075a04ad9d34f3ba47eaad97cd274e6 : SUCCESS in 18m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4fa60c569e655d9c55e2bb4ea206c2e59e813613","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 17:48:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f15b01b0265f398b4ffbf64256343ef02f96d03c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-30 07:05:41.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 08a3cf6b58451d6de698559cf945fc7b53ceaa6b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"062c866e29aceef50548ed50c8496c447bfcbf69","revisions":{"062c866e29aceef50548ed50c8496c447bfcbf69":{"kind":"REWORK","_number":1,"created":"2020-05-29 08:35:56.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/93/731693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/93/731693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/93/731693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/93/731693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/93/731693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/93/731693/1"}}},"commit":{"parents":[{"commit":"e9de5edbfce477a913b109a7c0492ec8cba1e640","subject":"Remove unnecessary puppet-rabbitmq from dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/e9de5edbfce477a913b109a7c0492ec8cba1e640"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-29 08:33:58.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-29 08:33:58.000000000","tz":120},"subject":"Chain barbican_api_paste_ini to config anchor","message":"Chain barbican_api_paste_ini to config anchor\n\nMake sure the barbican_api_paste_ini resources\nis performed in the config anchor so that the\npackages is installed already.\n\nWe got his by this failure see error message\nhere [1].\n\n[1] http://paste.openstack.org/show/794135/\n\nChange-Id: I5de77441cd741b809762166347ec6bef04ba4f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/062c866e29aceef50548ed50c8496c447bfcbf69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/062c866e29aceef50548ed50c8496c447bfcbf69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
