)]}'
{"id":"openstack%2Fpuppet-barbican~731871","triplet_id":"openstack%2Fpuppet-barbican~master~Iaafc7f4b1499df471a35410228a7ad86938e94cc","project":"openstack/puppet-barbican","branch":"master","topic":"debian-uwsgi-not-gunicorn","hashtags":[],"change_id":"Iaafc7f4b1499df471a35410228a7ad86938e94cc","subject":"Debian: do not configure gunicorn-config.py","status":"MERGED","created":"2020-05-30 01:04:11.000000000","updated":"2020-05-31 15:12:27.000000000","submitted":"2020-05-31 15:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731871-1590937947613-c0d76a93","meta_rev_id":"2e53cdff2336f936d56c6e535d5aa34b129c78b1","_number":731871,"virtual_id_number":731871,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-31 15:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-31 14:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-05-31 08:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"date":"2020-05-30 10:34:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-05-31 14:11:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-05-30 10:34:51.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-30 10:34:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-30 10:34:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-31 08:19:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-31 14:11:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-31 15:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3d966bd9b9011b155e68dd69803786ee5244211","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-30 01:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"456814f26766687bfd03cfccc2265ae39b86828b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 02:10:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dbdfb1396d24123bbefcac842b126d2 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a4a1f45409c41f6bad594e0b55dc6cb : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8ecdf3469c443a08ec551aed12946cf : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/02281dec2ad1460c8cecd49edaa4d375 : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9056bd813888494bbfe60b412b4155ab : SUCCESS in 9m 19s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/354d4f4a11dc41f78f9b2a77c6478fa9 : SUCCESS in 14m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cff6f12625a646bdb0f9f7c56a1d84f2 : SUCCESS in 10m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84dc2b87231440c193cb4702ae2aa4f7 : SUCCESS in 1h 02m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9c64c84dc8c47f7b10e9eed72b36985 : SUCCESS in 1h 01m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9b3caf31880c418aac6373e4587a38e3 : SUCCESS in 24m 16s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/55675bbe5a3d4c0397432bc1f3579c54 : FAILURE in 13m 59s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a53155c4dce6464696bcf0c26ee9dac2 : FAILURE in 9m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f65dc242e30fba5ff93ec0e20e95bcf30a4a4eb9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-31 08:19:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee3153c5eef4ce5a3152764dd9ab041e701827a7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-31 14:11:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43359e57c0c1c6bc6b7ec1abe8066ad378ee3a5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 14:11:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a64c7575bfd26111752fbb2485a46899205409c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 15:12:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25fdead39691468fb7888eee06c9a37d : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/370e8b0485474a8880f70b27cf413edf : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d316180527484cc4ac3bf1c98d093aaa : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/331e90abfea14366a755e6323b066f69 : SUCCESS in 10m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f22b7a45fc4d4ffea577925f33d462df : SUCCESS in 12m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40081d64884949ca9140185ad0cfa0a9 : SUCCESS in 10m 07s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d8f86da3fb74ac6b7fb091563abdcbe : SUCCESS in 51m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b2ba9a206da44c5f91d23298c2bfebc8 : SUCCESS in 58m 41s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/29b61739050649c0a2a991ab6a457469 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"62f291aa4a3e4dfac872666d33193beb99228366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 15:12:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60d3b6826668c96f538aaea4db72b93abccad078","revisions":{"60d3b6826668c96f538aaea4db72b93abccad078":{"kind":"REWORK","_number":1,"created":"2020-05-30 01:04:11.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/71/731871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/71/731871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/71/731871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/71/731871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/71/731871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/71/731871/1"}}},"commit":{"parents":[{"commit":"062c866e29aceef50548ed50c8496c447bfcbf69","subject":"Chain barbican_api_paste_ini to config anchor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/062c866e29aceef50548ed50c8496c447bfcbf69"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-30 01:02:49.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-30 01:02:49.000000000","tz":120},"subject":"Debian: do not configure gunicorn-config.py","message":"Debian: do not configure gunicorn-config.py\n\nDebian isn\u0027t using gunicorn for Barbican, but UWSGI.\nWithout this patch, this module simply wouldn\u0027t work\nfor Debian.\n\nChange-Id: Iaafc7f4b1499df471a35410228a7ad86938e94cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/60d3b6826668c96f538aaea4db72b93abccad078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/60d3b6826668c96f538aaea4db72b93abccad078"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
