)]}'
{"id":"openstack%2Fpuppet-barbican~731882","triplet_id":"openstack%2Fpuppet-barbican~stable%2Fussuri~I5de77441cd741b809762166347ec6bef04ba4f4c","project":"openstack/puppet-barbican","branch":"stable/ussuri","hashtags":[],"change_id":"I5de77441cd741b809762166347ec6bef04ba4f4c","subject":"Chain barbican_api_paste_ini to config anchor","status":"MERGED","created":"2020-05-30 07:05:41.000000000","updated":"2020-05-30 09:10:48.000000000","submitted":"2020-05-30 09:10:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731882-1590829848016-2d12829e","meta_rev_id":"aad13acf764acdbe1a5ceb2ac9df44e379810a15","_number":731882,"virtual_id_number":731882,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-30 09:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-05-30 08:02:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2020-05-30 08:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-30 09:10:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a777f9676c026f461f3dc3ed93681f127c9d8ec0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-30 07:05:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"15020d02abafd4c3c2b48328aeda423d2ce3d09a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-30 07:05:59.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit aa6afa0124198139b57741bd35d3fccf24b9fae5","accounts_in_message":[],"_revision_number":1},{"id":"ef1548fd4c632a1a6ffbf926c0db6855c2ea5f8e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-30 08:02:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b02153149e8b791c0228f7a8dcfc84506216ea6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 08:10:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bbc9ab18b0348058623d428d5e85e9e : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0057afed2134b76b865c0c8e62da7bb : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7a37f7888654364aaacf57007f2a6dd : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/811fb263fdef45e9aa9b8802d34c319c : SUCCESS in 16m 02s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/098c62ea985b4bd99cd9bc526786d7a6 : SUCCESS in 16m 40s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32e5208c68fb4bed94760f32c871a3bf : SUCCESS in 11m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94932fab05dc4953bb09c75d102b2c01 : SUCCESS in 15m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b9e8ae5a0ad40ce9592f1a16448eb88 : SUCCESS in 55m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9319f9a20ceb4719a87d42b38780665b : SUCCESS in 1h 03m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/89412009e6934aaaa56db6f090604746 : SUCCESS in 23m 32s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/056d52f4b83c45df8d90ed69efcdc1a5 : FAILURE in 13m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91b06ce6d15044bebfc3804ca1c75d20 : FAILURE in 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f35894c627ce3cae9c2d6c5df28469bc92fb729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 08:10:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec7f5987eccfaf1e5245c17774375d32265746b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 09:10:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b642c7cd8a2142709d1c496ecc560c93 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21b9d820330940b78430a18dfbb2b463 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7016616fbc7483a8a710d64c461f522 : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2ccd54a21b0e42e8b77ec7cf665bebfa : SUCCESS in 11m 14s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1e438fd23c8b474a9199a6ce8399da47 : SUCCESS in 10m 48s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63079cb1d2cf4c949419460ffdd7e218 : SUCCESS in 8m 49s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62db18b62fff4d1f843459e77d8558ad : SUCCESS in 50m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9b8a98e0dab24ccaa3b089848b93b087 : SUCCESS in 58m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3bfe56ba487f43eb8a2fc495ec32ddaa : SUCCESS in 19m 04s","accounts_in_message":[],"_revision_number":1},{"id":"aad13acf764acdbe1a5ceb2ac9df44e379810a15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 09:10:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08a3cf6b58451d6de698559cf945fc7b53ceaa6b","revisions":{"08a3cf6b58451d6de698559cf945fc7b53ceaa6b":{"kind":"REWORK","_number":1,"created":"2020-05-30 07:05:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/82/731882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/82/731882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/82/731882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/82/731882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/82/731882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/82/731882/1"}}},"commit":{"parents":[{"commit":"a1f1c9fc5acdde46ebfd8f7efc6ebf5866b83baf","subject":"Switch to Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a1f1c9fc5acdde46ebfd8f7efc6ebf5866b83baf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-29 08:33:58.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-30 07:05:41.000000000","tz":0},"subject":"Chain barbican_api_paste_ini to config anchor","message":"Chain barbican_api_paste_ini to config anchor\n\nMake sure the barbican_api_paste_ini resources\nis performed in the config anchor so that the\npackages is installed already.\n\nWe got his by this failure see error message\nhere [1].\n\n[1] http://paste.openstack.org/show/794135/\n\nChange-Id: I5de77441cd741b809762166347ec6bef04ba4f4c\n(cherry picked from commit 062c866e29aceef50548ed50c8496c447bfcbf69)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/08a3cf6b58451d6de698559cf945fc7b53ceaa6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/08a3cf6b58451d6de698559cf945fc7b53ceaa6b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
