)]}'
{"id":"openstack%2Fpuppet-barbican~740126","triplet_id":"openstack%2Fpuppet-barbican~master~I5068505afbaf57e66d28f7cf472ab09bb8355f04","project":"openstack/puppet-barbican","branch":"master","topic":"at-interface","hashtags":[],"change_id":"I5068505afbaf57e66d28f7cf472ab09bb8355f04","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-07-08 23:19:52.000000000","updated":"2020-07-09 09:01:02.000000000","submitted":"2020-07-09 08:59:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740126-1594285172924-447c4ef9","meta_rev_id":"a9f1b1cdca48d5d6af21e87baf7702f673416465","_number":740126,"virtual_id_number":740126,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-09 08:59:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-07-09 02:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-07-09 07:46:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-07-09 07:46:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 02:02:16.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-07-09 07:46:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-09 08:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0e7c4fce048b49243d19ff252f2d77f860d082f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-08 23:19:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a18fe2e0a8aafb35b33e51d396073ba716928fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 00:28:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40514c7556fe42ca8e504c477e8d48f1 : SUCCESS in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a8aa1674fa742f498c3c0516be7d1c7 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd32fe6464f341bbb9973a8c26e0e7ae : SUCCESS in 6m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6166b32d70bd454baece6a1be536afe5 : SUCCESS in 12m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4110f647767140c891103c0ce69cb81a : SUCCESS in 14m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70388452f82442798d04068085c511e9 : SUCCESS in 10m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56979a70ab9d4f56baedd568da760216 : SUCCESS in 10m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cd26f0126ae4cd79e71561972f62b5f : FAILURE in 52m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b6e1789a6d34c9d81cae7f2bf0c46cd : SUCCESS in 1h 07m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/69297b8354124c2babee5c44bc5dbfff : SUCCESS in 17m 50s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/61e72718baaf4db1a2ad8d5cb5b5e4b2 : FAILURE in 11m 42s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/842cf887b6e948d4bf769bd831edad08 : FAILURE in 9m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e33b8592246e4f8286ba59924ca2fffb : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e8ab97cdaa6126ec27a7f5e0feaaf7ea5082cfec","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-07-09 02:02:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c53bf4fa5083e578920d7a42825f3fe1edffc7c7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-09 07:46:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1867072e344920471b3a0890c88873e79a5d4919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 07:46:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b0e7f4f526a7a21abe8df3bdfe7cc307a2735ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:59:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c62973e40a45b940580c61bf4f3502f8415bf5e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:59:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22995c41442046e083942b476f873519 : SUCCESS in 5m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f02401977c7f44ba90cd4c0056d116ae : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/546d95e2c233441694c745459e89a1ce : SUCCESS in 6m 08s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9ad50ecccce44179a83f7165cd5ee574 : SUCCESS in 12m 51s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/14ca5b8e1a074cb4883bda9f332d5409 : SUCCESS in 16m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a40412a42a54ae092b8e932de1d5dfc : SUCCESS in 14m 10s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fb93fb2a9814e9ba7bff1edc1264ad1 : FAILURE in 50m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2a0b4a17d58f4fce97a786ef720bea58 : SUCCESS in 1h 05m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d6f3783cdaef4338b8f79d2b840a96e8 : SUCCESS in 20m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15715d9d1e1445b89cfd344f5e03f1fd : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"20f782aed4ade6fae5354dd73431c88af0a12ca1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:01:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8d96b2359b74657a55f523262dea70f : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"000d145608baad5daffcb2b80863b45b3200e435","revisions":{"000d145608baad5daffcb2b80863b45b3200e435":{"kind":"REWORK","_number":1,"created":"2020-07-08 23:19:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/740126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/26/740126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/26/740126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/26/740126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/26/740126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/26/740126/1"}}},"commit":{"parents":[{"commit":"bbd38354a0b25c5e32fc3c99fb6ace2c9a87f0c7","subject":"Prepare Victoria M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/bbd38354a0b25c5e32fc3c99fb6ace2c9a87f0c7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 23:19:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 23:19:30.000000000","tz":540},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I5068505afbaf57e66d28f7cf472ab09bb8355f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/000d145608baad5daffcb2b80863b45b3200e435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/000d145608baad5daffcb2b80863b45b3200e435"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
