)]}'
{"id":"openstack%2Fpuppet-barbican~746523","triplet_id":"openstack%2Fpuppet-barbican~stable%2Ftrain~I3baa7af29970b4da25ff67a54e473c32ee6ad40f","project":"openstack/puppet-barbican","branch":"stable/train","hashtags":[],"change_id":"I3baa7af29970b4da25ff67a54e473c32ee6ad40f","subject":"Expose vhost_custom_fragment","status":"MERGED","created":"2020-08-17 14:50:01.000000000","updated":"2020-08-18 08:08:30.000000000","submitted":"2020-08-18 08:08:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746523-1597738110291-44e8c23f","meta_rev_id":"ffd4f3b9031020c3d657511956ec56c6253f599a","_number":746523,"virtual_id_number":746523,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 08:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-17 21:52:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-18 07:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-18 07:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 21:52:40.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-18 07:01:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-18 08:08:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c95d99d35a86d7a3316a166af1ecb28d5301efb2","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-17 14:50:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99f122c1a78169312964dea8755ef17bb3263b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:55:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a61039e0de648ba9130e2c07e131ac1 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ed1b67e677c433598780dd71db4e27d : SUCCESS in 5m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75c0c4f3f11b4f30a6c8211a5d25e6a6 : SUCCESS in 7m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/04a943e9c06e480fa221c6a4bfba7fd5 : SUCCESS in 13m 06s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/80d104dd4e6a42e8af863347314beebc : SUCCESS in 13m 34s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/ad536f28669e44f0b350354fe9dcce6a : SUCCESS in 15m 29s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36f042fd8f464b038271a203f53b2392 : SUCCESS in 11m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4423cec082144b95922c2e99d6df6794 : SUCCESS in 13m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5c536a7d37c4ba1bf2bc70d2d1388fb : SUCCESS in 48m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a913f54a6c2f4d7e94b62fd3864a0b8e : SUCCESS in 59m 01s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c8e83637d07641118127c5974e40acbc : FAILURE in 15m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/df54a7b95970436b86d6e23e4e08ad34 : SUCCESS in 17m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a4be76fed1b4b6989673fca75bd36ab : FAILURE in 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9a4cb06216e539ecd74ee5d808d073757700bfb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 21:52:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4191bdc481a20335242eb22163f10aae047c3c9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-18 07:01:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b57181affccc41499dadddf25f5866f7fa239fc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28adbd171f4090ef6a3eaf625419fbeb57f6b4a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:08:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ffd4f3b9031020c3d657511956ec56c6253f599a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:08:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db1b663aa1354a3a988126f1dcbf3041 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3583ed5b02a244b28a87e0534b29a65a : SUCCESS in 5m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b4c56c457b245a5ab8f90ff8cb2dbe6 : SUCCESS in 5m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/09bf7da71709406fae7a009b276e3f2d : SUCCESS in 17m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c539913e5e1945caaa2495bf0644ea89 : SUCCESS in 13m 10s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/457be8f37294467ba7208e98a11faf6e : SUCCESS in 43m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cc25539ae2a74901b6ff80f4184aa267 : SUCCESS in 59m 34s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b6f493e0a77540b38a01878b4a957fde : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da146d53877db069a521d49b79f70e2827e8caad","revisions":{"da146d53877db069a521d49b79f70e2827e8caad":{"kind":"REWORK","_number":1,"created":"2020-08-17 14:50:01.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/23/746523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/23/746523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/23/746523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/23/746523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/23/746523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/23/746523/1"}}},"commit":{"parents":[{"commit":"aa6afa0124198139b57741bd35d3fccf24b9fae5","subject":"Chain barbican_api_paste_ini to config anchor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/aa6afa0124198139b57741bd35d3fccf24b9fae5"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-13 14:32:07.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-17 14:49:16.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Barbican service.\n\nConflicts:\n    manifests/wsgi/apache.pp\n\nChange-Id: I3baa7af29970b4da25ff67a54e473c32ee6ad40f\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n(cherry picked from commit 257f929622f9e8e0791930fca3b1069accf03c50)\n(cherry picked from commit 288dced2d324e7e68af86f1e95603a8b5c4ae5bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/da146d53877db069a521d49b79f70e2827e8caad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/da146d53877db069a521d49b79f70e2827e8caad"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
