)]}'
{"id":"openstack%2Fpuppet-barbican~756102","triplet_id":"openstack%2Fpuppet-barbican~master~I4c3ba63dee6ec9081ec083ce45620cd6966e6c4b","project":"openstack/puppet-barbican","branch":"master","topic":"config-lint","hashtags":[],"change_id":"I4c3ba63dee6ec9081ec083ce45620cd6966e6c4b","subject":"Update comment to fix lint","status":"MERGED","created":"2020-10-05 15:13:00.000000000","updated":"2020-10-12 02:36:31.000000000","submitted":"2020-10-12 02:36:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756102-1602470191726-75d80b9d","meta_rev_id":"65fb4d29c7da30a0d007cf1036442671c0dce51b","_number":756102,"virtual_id_number":756102,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 02:36:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-12 01:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-06 10:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-10-10 15:32:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-12 01:23:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2020-10-10 03:14:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 10:23:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-10 15:32:06.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2020-10-12 01:23:14.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-12 02:36:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"892835601104506b639afd92035d2c4f87dc8a7d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 15:13:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"119658d9e2d41ebf8f59f8206365f26f79ef689a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-05 16:22:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73ece45a335fd71ea164cd85abb26aba18d467c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 18:13:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d4dad0104a14f0d8050add9109ad9ae : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56e06b785bfd45e58036903d94be34ff : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6ec5753172b1404e91eb62426fac15ab : SUCCESS in 9m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/553493838d984a43b6319e4acac8bdfb : SUCCESS in 8m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66eae8be09254d71ae08686ae34d53e6 : SUCCESS in 11m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea868a46acbc4051a3572ffc05acd83b : FAILURE in 1h 25m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/09e3a6d1b61541d5acbcae0a6cfff715 : SUCCESS in 1h 29m 39s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/3b07ee2b46ec4736a3a962bfc432152a : SUCCESS in 18m 26s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dda49bf3f1ee44eeb9d8287a860b13a7 : SUCCESS in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"411cef03917682d3b767daf765e568d49fa40218","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-06 10:23:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d912d2b27cb9516937ff8ce212c9d9027c2f373d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-06 12:04:42.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"5fc62e018719667658d8394647efe8ecd88f8e8a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-10 03:14:17.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"bf41cfff3cd53fab641c2b76c7f5bd5ec64082cd","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2020-10-10 15:32:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"176d526f797e0787f77e39a68bcf390291468663","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-12 01:23:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"239357515e5de74d1ae50ab66bb20382efa40bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 01:23:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dfbfd1dcf234d876b1adfa9275a1ebf1ff1f62ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 02:36:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dd1972053ba8b0f55d0a7f12a246b715b6690430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 02:36:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c8f704ebdf24bf89fd953d542e64bcb : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e23f20c8276f4f338073e89c3e648504 : SUCCESS in 5m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/aaa6532076fb41939dab5f489970cf5d : SUCCESS in 18m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/699dcf570b0e447ba1cb41be48220d87 : SUCCESS in 7m 14s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/829d5fe71991471596bc3bd4d90ee6d8 : FAILURE in 32m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/267a0e6e6fc6413087e3ef083d45e4df : SUCCESS in 1h 11m 13s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ac8d8bb61c2849f291461a86562849d1 : SUCCESS in 13m 46s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef0b9a42efb44095899e1ff67c211f97 : SUCCESS in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3b2cde6e33fe2183d35bb0036a3bcbbda80917a3","revisions":{"a7eb87d17a6b4d845f13ddbcd7c8427f5ed3b059":{"kind":"REWORK","_number":1,"created":"2020-10-05 15:13:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/756102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/02/756102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/1"}}},"commit":{"parents":[{"commit":"5607e63e785bd5e2c59bd853575196f1c7a5fdd0","subject":"Merge \"Do not validate database_connection format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/5607e63e785bd5e2c59bd853575196f1c7a5fdd0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 15:12:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 15:12:45.000000000","tz":540},"subject":"Update comment to fix lint","message":"Update comment to fix lint\n\nmanifests/config.pp:7:WARNING: No matching class parameter for\ndocumentation of barbican::config::xxx_config\n\nChange-Id: I4c3ba63dee6ec9081ec083ce45620cd6966e6c4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a7eb87d17a6b4d845f13ddbcd7c8427f5ed3b059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a7eb87d17a6b4d845f13ddbcd7c8427f5ed3b059"}]},"branch":"refs/heads/master"},"3b2cde6e33fe2183d35bb0036a3bcbbda80917a3":{"kind":"REWORK","_number":2,"created":"2020-10-05 16:22:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/756102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/02/756102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/02/756102/2"}}},"commit":{"parents":[{"commit":"5607e63e785bd5e2c59bd853575196f1c7a5fdd0","subject":"Merge \"Do not validate database_connection format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/5607e63e785bd5e2c59bd853575196f1c7a5fdd0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 15:12:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-05 16:22:04.000000000","tz":540},"subject":"Update comment to fix lint","message":"Update comment to fix lint\n\nmanifests/config.pp:7:WARNING: No matching class parameter for\ndocumentation of barbican::config::xxx_config\nmanifests/init.pp:7:WARNING: No matching class parameter for\ndocumentation of barbican::sample_parameter\n\nChange-Id: I4c3ba63dee6ec9081ec083ce45620cd6966e6c4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/3b2cde6e33fe2183d35bb0036a3bcbbda80917a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/3b2cde6e33fe2183d35bb0036a3bcbbda80917a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
