)]}'
{"id":"openstack%2Fpuppet-barbican~765274","triplet_id":"openstack%2Fpuppet-barbican~stable%2Fussuri~I5068505afbaf57e66d28f7cf472ab09bb8355f04","project":"openstack/puppet-barbican","branch":"stable/ussuri","topic":"at-interface-stable/ussuri","hashtags":[],"change_id":"I5068505afbaf57e66d28f7cf472ab09bb8355f04","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-12-03 06:35:49.000000000","updated":"2020-12-04 07:58:18.000000000","submitted":"2020-12-04 07:57:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765274-at-interface-stable/ussuri","cherry_pick_of_change":740126,"cherry_pick_of_patch_set":1,"meta_rev_id":"655657e8a53f6540921b291155ce090557f4fb33","_number":765274,"virtual_id_number":765274,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 07:57:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"recommended":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"all":[{"value":1,"date":"2020-12-03 12:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":2,"date":"2020-12-04 06:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-12-03 21:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":1,"date":"2020-12-04 06:53:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 08:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 12:07:58.000000000","updated_by":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"reviewer":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"state":"REVIEWER"},{"updated":"2020-12-03 21:57:36.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-12-04 06:53:22.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"b54ee6ce1f1886c3b245b917ffdbb459d4a8faee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-03 06:35:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"002d479b41efd2f6e2e071f548f1a0bc428dbd81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-03 06:43:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"60cdbb7f979a45d7bed9b610c95848619a922b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 08:56:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4faede2356d4351949b8e021de642ea : SUCCESS in 3m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cc3166491fa4612b7c8dfbcc8d6f31f : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b5457a3debb44459f100c3b41978746 : SUCCESS in 3m 56s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2ecc919215804d70896723346937e6a0 : SUCCESS in 9m 02s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1ebb6aba2cd144b289b52ae639f66ba1 : SUCCESS in 9m 53s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc4d8ad9e32c42e89f5a88ea8f7b0c88 : SUCCESS in 10m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4be7c721695d40a3832a1816ed3a4713 : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e54ff2e2c5214509b823c044cd337130 : SUCCESS in 1h 00m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/59878bc132af40a8af34f4b9abee4a92 : SUCCESS in 1h 05m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4d1b204b68c74f06bb43d144e7e4aa9c : SUCCESS in 19m 13s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1162295a4e7944b7a02c53aefb9c6468 : SUCCESS in 13m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f28b28a5134f404fadbbeac6b2e3dbcc : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8b390ff9cb2ad3c83f4ee93531678cc097b5bfe5","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2020-12-03 12:07:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e3e20a9341df9a2dca134149ad70badfa8247c0f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-03 21:57:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6b70a4d7d1b131e7229098016b9fe1f813328ce7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-12-04 06:53:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7747182599f381953b16000497ba8ca63fbac40a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:53:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1fbdc5e47d1cd7631fad80ce60c1155add7122b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:57:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cc9ae906489408b8ff8ccb645a157db : SUCCESS in 7m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/645fbe7608ee4093871bc61729264d3a : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cc724a9e6e045dba8b323b4602c883a : SUCCESS in 3m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/988af31c47e74e898216576bcfebcc2b : SUCCESS in 10m 27s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/470e1ac1e5ad462eaa37d194231dd584 : SUCCESS in 8m 20s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff558448e55747a78567ec3882ada3ab : SUCCESS in 6m 34s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aacf2bc84d445e7b1b309273a77306b : SUCCESS in 52m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/867b52e1554242b18a2b3a4167e01042 : SUCCESS in 48m 16s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/5399afaaa11045eca82766b45529f008 : SUCCESS in 20m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cff2920a5964179bdb0eaf3e73ae38a : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":2},{"id":"042fdcb6c056a8ae4445a9c144fac1f731211206","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:57:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"655657e8a53f6540921b291155ce090557f4fb33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:58:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7108211cde074346b765ecfd83f5735a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72c6aa190480d99970549125c646f3a8cacd5d7e","revisions":{"7ba8e0d55dd9054380149d58cc6a4910323dc1f5":{"kind":"REWORK","_number":1,"created":"2020-12-03 06:35:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/765274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/74/765274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/1"}}},"commit":{"parents":[{"commit":"db83a4a10953c8f96b7d269586b3e5d2c36bfaaa","subject":"Add p11_crypto_plugin token options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/db83a4a10953c8f96b7d269586b3e5d2c36bfaaa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 23:19:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-03 06:35:49.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I5068505afbaf57e66d28f7cf472ab09bb8355f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7ba8e0d55dd9054380149d58cc6a4910323dc1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7ba8e0d55dd9054380149d58cc6a4910323dc1f5"}]},"branch":"refs/heads/stable/ussuri"},"72c6aa190480d99970549125c646f3a8cacd5d7e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-03 06:43:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/765274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/74/765274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/74/765274/2"}}},"commit":{"parents":[{"commit":"db83a4a10953c8f96b7d269586b3e5d2c36bfaaa","subject":"Add p11_crypto_plugin token options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/db83a4a10953c8f96b7d269586b3e5d2c36bfaaa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 23:19:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-03 06:43:18.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I5068505afbaf57e66d28f7cf472ab09bb8355f04\n(cherry picked from commit 000d145608baad5daffcb2b80863b45b3200e435)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/72c6aa190480d99970549125c646f3a8cacd5d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/72c6aa190480d99970549125c646f3a8cacd5d7e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
