)]}'
{"id":"openstack%2Fpuppet-barbican~773270","triplet_id":"openstack%2Fpuppet-barbican~master~I7cbd1e6f2b5262e158a79153f049a0254f8f1087","project":"openstack/puppet-barbican","branch":"master","topic":"healthcheck","hashtags":[],"change_id":"I7cbd1e6f2b5262e158a79153f049a0254f8f1087","subject":"Add support for healthcheck middleware options","status":"MERGED","created":"2021-01-31 23:06:47.000000000","updated":"2021-02-24 02:31:54.000000000","submitted":"2021-02-24 02:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773270-healthcheck","meta_rev_id":"22f989e7ca806a19ce2d7c75490aa62e7098030d","_number":773270,"virtual_id_number":773270,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 02:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-02-01 08:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-02-23 17:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-02-23 17:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 00:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 08:35:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-02-23 17:48:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e7f8eebc47ea39a77d8680b9308f44fb185e9a76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-31 23:06:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7de23d2dfc75edb5bd53adf940678485d27cdd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-31 23:12:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98b3137e92b863e642c4681ced9a812e3fe059de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 00:49:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7ed5021e05e4abaaeaa0ae321aa58bd : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3be8c23983047fbb557c5c560505077 : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/966741ba772b4ec39c360fcfc8ad0096 : FAILURE in 7m 10s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/79da69e97749486d93eed71ad61e8e63 : FAILURE in 7m 51s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bfc03f46cd84a77bce64aa57cf3bc63 : FAILURE in 8m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6385efd9abc1436197f7406c5d7c49be : FAILURE in 6m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ea83b5103a84739a9d853d42b7a7dfa : SUCCESS in 54m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e01f9953024d449bbcb7d0bfdfafb3c9 : SUCCESS in 1h 33m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f6729c144f34ca4a89736cf356d170d : SUCCESS in 1h 35m 21s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6232eaaa6e13480d97c4ff643c5cb55d : SUCCESS in 14m 06s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c35efc5cdb240d79e76529257e1344c : SUCCESS in 10m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eebf11c673a46d1a89c36ff266c9885 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3a1259530711d09be559ea1d15998b77474c0f0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-01 01:33:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"399ce54a3197934e80c5454c6974286d4361aa54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 03:16:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75590cdbd0824203aa56462dd06dc9e1 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/703357dcb8314f0fa3306068718bb42c : SUCCESS in 2m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/dfa7699e946c429abad596923929c0c6 : SUCCESS in 8m 56s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d7c9fa689074036a7c1bbbd4ce922c0 : SUCCESS in 8m 42s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a00964bd6e5c4998bdf8d21a4b8010d8 : SUCCESS in 6m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc0d506c95fa46a1bc78f5a812fd3c30 : SUCCESS in 6m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a59a73c85adf41aa84d88958f349afc6 : SUCCESS in 56m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ddd3099450fc407997b45a85bc1ebaf1 : SUCCESS in 1h 37m 08s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6fba6dab6a844b29bc07c3b2942a19df : SUCCESS in 1h 41m 55s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/831576b323b14e24b84b6fed19953a8f : SUCCESS in 16m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3e2169574d047a89360a09774963131 : SUCCESS in 12m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a371317bd53b452faed58172f7de35c9 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"70c0c18d2ccaec983ceb744d269ca6e1869ae027","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-01 08:35:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e7197e962328da84207b3b4bbd8d5f12e92a1e99","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 17:48:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6d3699ccd05b28870c511643284ced6c396c7bfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:48:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9ddd772c40639fafab49bcfdbf981a419be1473d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:55:12.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad91574dcd604254bdce28b0129d04ee : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45f0ec9abbea42fca8541019b989fa21 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/68d72878de714396a54c27d90db5a0a6 : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/beb6d0c84b134dd8a0360a4737885c24 : SUCCESS in 7m 26s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d3c96688f774346bb229f47b4f3c5c6 : FAILURE in 35m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9672b42903cf499284edd10850308c67 : TIMED_OUT in 2h 01m 04s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6415d7ec4ad54c8a83df44fa1490d4d5 : SUCCESS in 17m 36s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50c6c509b19a4324b8df303a278d4820 : SUCCESS in 19m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a05bce4bf0d4694b12fb94ad412cba9 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"d387a162c5b8b4c36c63441373599a949a412e11","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-23 22:44:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f11e4407e39d2c5506fa25d5cf345c340ec23750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 00:26:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f12d8b0318349329fefa5e71db075ec : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3599748a9be42cebeb921f876211892 : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/37e8d11ee8e349d9a84eebe261433d8f : SUCCESS in 9m 22s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ee1fe813ca9411e9c1ba2f89d9907d5 : SUCCESS in 9m 22s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05284c81cdf949f8b8abb18b224d29d5 : SUCCESS in 7m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b8bb218257b4d79adf75cfeb022ebf1 : SUCCESS in 7m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c2e0ebe328c48d0898efa667d64d262 : FAILURE in 37m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c34c16421a944fb3ba9d481857ca2771 : SUCCESS in 1h 30m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e3bc41b85574e54a5774fd29cef25a8 : SUCCESS in 1h 31m 30s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/34fd5fd5aa0a45c6a8693bc7d1462e91 : SUCCESS in 20m 41s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f60068f0fe5f418c82a112842dfea21f : SUCCESS in 19m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d11e119bbe7f4dbbb2540fb028b8264d : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f2e179decf99bcd0cf37aa6043bf733e3ad3bdd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 00:26:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6ad05287b3c7fd17b89ba4f56dd787d6c0875603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 02:30:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9ec8afff5f3452b990ee1a29bda0056 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9f02b6214e94579a3163cd4fab2794c : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0310bc472b1d4c27950a42312b5155fe : SUCCESS in 13m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c01d80e78ae3448c8a17daeefd8c86f3 : SUCCESS in 8m 14s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d84a5f7af7be4e288634b32bf43085e3 : FAILURE in 37m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6bb4b7d0cce74c288060f51fc21d403f : SUCCESS in 1h 59m 11s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d5cc04e7d82d4050a7cae219a7aec548 : SUCCESS in 23m 12s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/233443778c9e488bb3b4739fc6b05523 : SUCCESS in 13m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbb820ad9b9b434fb69798e2dbfcdddb : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4f746b692009b62ec4a4294b52f4e66dd977db72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 02:30:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"22f989e7ca806a19ce2d7c75490aa62e7098030d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 02:31:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2796ce7dbfc1418294ff1bf61af9944d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"81135f6835182880961d54366b2d3014ff20a306","revisions":{"1a51769fd8e8f909492e61681eb456b75644970f":{"kind":"REWORK","_number":1,"created":"2021-01-31 23:06:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/773270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/70/773270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/1"}}},"commit":{"parents":[{"commit":"de19fdedf7611bf86ab8fc9fe53bcec3cc33e1ac","subject":"Prepare Wallaby M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/de19fdedf7611bf86ab8fc9fe53bcec3cc33e1ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-31 23:06:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-31 23:06:33.000000000","tz":540},"subject":"Add support for healthcheck middleware options","message":"Add support for healthcheck middleware options\n\nDepends-on: https://review.opendev.org/772202\nChange-Id: I7cbd1e6f2b5262e158a79153f049a0254f8f1087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1a51769fd8e8f909492e61681eb456b75644970f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1a51769fd8e8f909492e61681eb456b75644970f"}]},"branch":"refs/heads/master"},"48f0d03b22d4bd20a8d0409765a675c26792c18d":{"kind":"REWORK","_number":2,"created":"2021-01-31 23:12:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/773270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/70/773270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/2"}}},"commit":{"parents":[{"commit":"de19fdedf7611bf86ab8fc9fe53bcec3cc33e1ac","subject":"Prepare Wallaby M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/de19fdedf7611bf86ab8fc9fe53bcec3cc33e1ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-31 23:06:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-31 23:12:09.000000000","tz":540},"subject":"Add support for healthcheck middleware options","message":"Add support for healthcheck middleware options\n\nDepends-on: https://review.opendev.org/772202\nChange-Id: I7cbd1e6f2b5262e158a79153f049a0254f8f1087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/48f0d03b22d4bd20a8d0409765a675c26792c18d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/48f0d03b22d4bd20a8d0409765a675c26792c18d"}]},"branch":"refs/heads/master"},"81135f6835182880961d54366b2d3014ff20a306":{"kind":"REWORK","_number":3,"created":"2021-02-01 01:33:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/773270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/70/773270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/70/773270/3"}}},"commit":{"parents":[{"commit":"de19fdedf7611bf86ab8fc9fe53bcec3cc33e1ac","subject":"Prepare Wallaby M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/de19fdedf7611bf86ab8fc9fe53bcec3cc33e1ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-31 23:06:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-01 01:32:59.000000000","tz":540},"subject":"Add support for healthcheck middleware options","message":"Add support for healthcheck middleware options\n\nDepends-on: https://review.opendev.org/772202\nChange-Id: I7cbd1e6f2b5262e158a79153f049a0254f8f1087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/81135f6835182880961d54366b2d3014ff20a306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/81135f6835182880961d54366b2d3014ff20a306"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
