)]}'
{"id":"openstack%2Fpuppet-barbican~778065","triplet_id":"openstack%2Fpuppet-barbican~master~Iba7013dd6e1b1e4650b25cd4dd8dc1f355ceb538","project":"openstack/puppet-barbican","branch":"master","hashtags":[],"change_id":"Iba7013dd6e1b1e4650b25cd4dd8dc1f355ceb538","subject":"Add new options to PKCS#11 backend","status":"MERGED","created":"2021-03-01 20:52:22.000000000","updated":"2021-03-16 14:53:11.000000000","submitted":"2021-03-16 14:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"778065","meta_rev_id":"2b5854eeeb065b9957863a0baa6d43354ae78dc5","_number":778065,"virtual_id_number":778065,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-16 14:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2021-03-16 13:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-03-09 16:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-03-10 21:43:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2021-03-16 13:16:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 22:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 20:09:52.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-02 20:09:52.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-03-09 16:42:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-10 21:44:07.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-03-10 21:44:07.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-03-15 14:34:31.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"bb750e952bdcf02203a25dfe84908f8fa6796da3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-01 20:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f02cfd20b395b05b3207beaeb63318811f4de31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 22:37:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a3be27cec6f4f278b3ccbb8bd2a4a9e : FAILURE in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cef86c8628b141fa9d5a3e8790a3c12c : FAILURE in 5m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9d90143fc49e494892f6d17818266c23 : FAILURE in 10m 21s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a8d368cafa944498e34c53321393c61 : FAILURE in 10m 01s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e361682fd8fc4bd1bf405e1f790524fa : FAILURE in 6m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bf2bdb74cef48f6922220360b3ab130 : FAILURE in 5m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c54c13ec21d440482ecf0a347b37678 : FAILURE in 38m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f47d1b6ab072493ca2783e3fce4e2555 : SUCCESS in 1h 38m 25s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/06b4dde5a2a7492484c9af5b3d4baa85 : SUCCESS in 1h 36m 10s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e18ebd049b174872acc4a2ffbe5ed7a2 : FAILURE in 8m 34s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fba19a824234ccaba1eee25a07e5a15 : FAILURE in 6m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f66a61c0fd2c45bc8ac2eae03b3680d0 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"19e50680e299b4fd2263215dd0c7df3a234b06dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-02 20:08:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec57d6a593178fd6dd6e9988d351b190b8c4383c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:44:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c4b30d71bdd45239e7009e6c68bbab4 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ec6481bc3294714b7fb403d729611f5 : SUCCESS in 3m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1214f196803f4895a8b51027309364a3 : FAILURE in 8m 08s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/6167936245234c4fb9794b3a90035aaa : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b05423a3af542519c0d86fee3be249b : FAILURE in 6m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf4471f85fcd448c92d9086720c55c5a : FAILURE in 6m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76acdf62b0794f17be39b9f3e0083328 : FAILURE in 35m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb9d2a05f9b748cfb039e46a5e35f1e1 : SUCCESS in 1h 17m 21s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b947363d20b4de1805ec62c027b1641 : SUCCESS in 1h 33m 53s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/71ccd7894fd8425998de931dc813df9c : SUCCESS in 16m 15s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/51d59c58b1b643b18370018ebc2127bc : SUCCESS in 16m 20s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2321b128fe147b4bda0e046a2d1c96c : SUCCESS in 11m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be47b265d1474330a15fb9fed6e4306a : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"033dc2f61ecf4620f166baab925afd87012d4474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-03 16:38:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5980f38ae4d22557a403a29bfb6335fcaf0ed819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 18:46:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd54c527880544ecb8b167ec96491d17 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58925e180ad542769610e074ac47cf5f : SUCCESS in 4m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4e88e9937dc443fbbd85a3fde1dfcb09 : SUCCESS in 11m 38s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/878172505936405a8f4b81fea3af5d8b : SUCCESS in 14m 47s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20a07f9544c8475c976a687e54f35d4f : SUCCESS in 10m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35b20758c6844be892fd23969bcfc3b2 : SUCCESS in 10m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8131a612d364352a02c494f5dac69f2 : FAILURE in 38m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/760e726a069f45f4afa8db75654ec5fc : SUCCESS in 1h 44m 32s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/62ebfa6b08724627865efd66de864001 : SUCCESS in 1h 32m 54s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/05391fabe5324b5eb6f0dcba357907e5 : SUCCESS in 24m 00s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2648b2695d74f5683f58f407c1d7852 : SUCCESS in 21m 03s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e11a5ae395141a1b2ea6e3aaa71d54d : SUCCESS in 17m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc3564d4378f4de7a71b9af30b4aac29 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":3},{"id":"889e042800cb018aaad9d907ee24aef97b0d40f5","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-03 19:45:20.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eb006b269244f68a944023c5a14ae6a7c6529754","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-03-03 19:53:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9003d09867672e6c91af1c2c3b25d84a94b17415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-04 13:39:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"95dae74a19ac631a526dc8fff5055954a0162289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 15:27:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e33e13d90c914771adbf2892fbab5a30 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20d9aa1e26b7460aaba277ba796f84a8 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8da498dacaf440d8a9f2fec113dc0f88 : SUCCESS in 9m 04s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/059a1681b2f1472c9dac72a30ddefe2d : SUCCESS in 8m 32s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ea86ad6054b4db1bf600bcb5844f06a : SUCCESS in 10m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2c815d9b8a64dcba38aed534cee0387 : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c79499173594431990cd708d0aa57cc0 : FAILURE in 43m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/10bb9999f9754822b536671201922d9e : FAILURE in 1h 44m 24s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9638ddb9a1194516969e9ca4f269f0d0 : SUCCESS in 1h 25m 29s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/df625b5a789c4931a2e0ae7838e231bd : SUCCESS in 21m 43s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5203fb98d98249c1a9122ff3ebb03cff : SUCCESS in 14m 44s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93a8ab74b16b43ac8e8ca0e3c6f083b6 : SUCCESS in 14m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a36935a7b2e40e89d051e624d71cde2 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"805d9cfd6f3b35171073b7fda8e803044a8704aa","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-08 14:56:18.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3dea840084038ec5127c6437eaba2500ed525bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:19:32.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"1064bd6cf0da514da17c4965552768596a50b918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-03-08 18:29:32.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0bbda0f2f84005fb4102bab9a116b6c7fa95c24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 20:24:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7115618c98b84aacb6c9820703c25806 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6649db7413f14b9091e42ff8090423d9 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/93664eeed0a34623be51448c2fb817b6 : SUCCESS in 9m 28s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ef090f678d249a4a83001080d1b470e : SUCCESS in 12m 20s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99fe747276cf4d92a6ac45e4b179d95a : SUCCESS in 8m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8afbd02dd8741a4ac064512402e8f77 : SUCCESS in 10m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c19b321ac5164f21be5bcc5b218c747c : FAILURE in 38m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d7ad90afad3d46cc9b0075d43a7edf60 : SUCCESS in 1h 32m 55s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/98095994421f4d6c989a4a67c4b90281 : SUCCESS in 1h 45m 37s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/643a99b8ad664e55b4617c5ac6192878 : SUCCESS in 19m 35s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a9c3040486045efa79005d4466c973c : SUCCESS in 15m 14s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86201efa92244f90be973b778b413d9c : SUCCESS in 13m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abbf9136247a493fa2fa264aa0652cd5 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":5},{"id":"9675068c29f5456857b472e98ee512ebf3ef52e4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-09 16:42:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8544874028181921efd72d5c30ef3dcc006d0b33","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-03-10 21:43:39.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9f22a3480833f1803c7e842b58260e5a300510eb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 13:16:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"678d50ea462ee8ace54a8bcd30ff4ebd06c8b05c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 13:16:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"59fb476723690f2cd7f5c3f2fb6d9686dbc5b3dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:44:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11aecf96980b45aaa720ef227e0e3f37 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de2bd8ca097748eb8dc59f312fc78e2f : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4213fbc8abe944f288583bc9c5db35c7 : SUCCESS in 12m 39s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4236c8c475a1404aa64c2eeab83dfd28 : SUCCESS in 7m 57s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dde3dbe5c4647ce876e2b59cc638bd2 : SUCCESS in 50m 37s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3b656e2d3fb34f6a9537ad1053b9b150 : SUCCESS in 1h 25m 38s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6c63251acb7d4a04af3bf888e1b1b89f : SUCCESS in 17m 23s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/de4d2445ea7648a78c321d9f63ad0079 : SUCCESS in 19m 49s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86b54f650ecb4db3b885227c0fc90a32 : SUCCESS in 14m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3b7c71cb3c447ae8d01e98126987104 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":5},{"id":"941a89de516e8587572d3f6e9a1112abbed0b5ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:44:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2b5854eeeb065b9957863a0baa6d43354ae78dc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:53:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60b46e0472624ff69a55be9ca8f7ecdc : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8685ed4a4cd3c886a00889bb150fbb3ba2620c8d","revisions":{"3ef14b6ff827923df7dcb1f171e25796bdce6437":{"kind":"REWORK","_number":1,"created":"2021-03-01 20:52:22.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/65/778065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/65/778065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/1"}}},"commit":{"parents":[{"commit":"81135f6835182880961d54366b2d3014ff20a306","subject":"Add support for healthcheck middleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/81135f6835182880961d54366b2d3014ff20a306"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-01 20:43:41.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-01 20:43:41.000000000","tz":-360},"subject":"Add token_labels option to PKCS#11 backend","message":"Add token_labels option to PKCS#11 backend\n\nThis patch adds support for a new option `token_labels` in the\nPCKS#11 backend.\n\nDepends-On: Ic2b85246c37e856c38cb47613313b19e653118de\nChange-Id: Iba7013dd6e1b1e4650b25cd4dd8dc1f355ceb538\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/3ef14b6ff827923df7dcb1f171e25796bdce6437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/3ef14b6ff827923df7dcb1f171e25796bdce6437"}]},"branch":"refs/heads/master"},"d54fd7688ac1f28b8553ba457c849d3f6dcfce0d":{"kind":"REWORK","_number":2,"created":"2021-03-02 20:08:02.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/65/778065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/65/778065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/2"}}},"commit":{"parents":[{"commit":"81135f6835182880961d54366b2d3014ff20a306","subject":"Add support for healthcheck middleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/81135f6835182880961d54366b2d3014ff20a306"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-01 20:43:41.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-02 20:07:08.000000000","tz":-360},"subject":"Add new options to PKCS#11 backend","message":"Add new options to PKCS#11 backend\n\nThis patch adds support for a couple of new new options in the\nPCKS#11 backends.  Namely \u0027token_labels\u0027 and \u0027os_locking_ok\u0027.\n\nDepends-On: Ic2b85246c37e856c38cb47613313b19e653118de\nChange-Id: Iba7013dd6e1b1e4650b25cd4dd8dc1f355ceb538\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d54fd7688ac1f28b8553ba457c849d3f6dcfce0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d54fd7688ac1f28b8553ba457c849d3f6dcfce0d"}]},"branch":"refs/heads/master"},"fae5ea540decffe6b96c3ed2a2f87159996aa3cb":{"kind":"REWORK","_number":3,"created":"2021-03-03 16:38:28.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/65/778065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/65/778065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/3"}}},"commit":{"parents":[{"commit":"81135f6835182880961d54366b2d3014ff20a306","subject":"Add support for healthcheck middleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/81135f6835182880961d54366b2d3014ff20a306"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-01 20:43:41.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-03 16:38:21.000000000","tz":-360},"subject":"Add new options to PKCS#11 backend","message":"Add new options to PKCS#11 backend\n\nThis patch adds support for a couple of new new options in the\nPCKS#11 backends.  Namely \u0027token_labels\u0027 and \u0027os_locking_ok\u0027.\n\nDepends-On: Ic2b85246c37e856c38cb47613313b19e653118de\nChange-Id: Iba7013dd6e1b1e4650b25cd4dd8dc1f355ceb538\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/fae5ea540decffe6b96c3ed2a2f87159996aa3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/fae5ea540decffe6b96c3ed2a2f87159996aa3cb"}]},"branch":"refs/heads/master"},"61cac077d3eee9daba3ebad5b8338e93f71c69bd":{"kind":"REWORK","_number":4,"created":"2021-03-04 13:39:57.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/65/778065/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/65/778065/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/4"}}},"commit":{"parents":[{"commit":"81135f6835182880961d54366b2d3014ff20a306","subject":"Add support for healthcheck middleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/81135f6835182880961d54366b2d3014ff20a306"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-01 20:43:41.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-04 13:39:51.000000000","tz":-360},"subject":"Add new options to PKCS#11 backend","message":"Add new options to PKCS#11 backend\n\nThis patch adds support for a couple of new new options in the\nPCKS#11 backends.  Namely \u0027token_labels\u0027 and \u0027os_locking_ok\u0027.\n\nDepends-On: Ic2b85246c37e856c38cb47613313b19e653118de\nChange-Id: Iba7013dd6e1b1e4650b25cd4dd8dc1f355ceb538\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/61cac077d3eee9daba3ebad5b8338e93f71c69bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/61cac077d3eee9daba3ebad5b8338e93f71c69bd"}]},"branch":"refs/heads/master"},"8685ed4a4cd3c886a00889bb150fbb3ba2620c8d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-08 18:29:32.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/65/778065/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/65/778065/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/65/778065/5"}}},"commit":{"parents":[{"commit":"81135f6835182880961d54366b2d3014ff20a306","subject":"Add support for healthcheck middleware options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/81135f6835182880961d54366b2d3014ff20a306"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-01 20:43:41.000000000","tz":-360},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-03-08 18:28:48.000000000","tz":0},"subject":"Add new options to PKCS#11 backend","message":"Add new options to PKCS#11 backend\n\nThis patch adds support for a couple of new new options in the\nPCKS#11 backends.  Namely \u0027token_labels\u0027 and \u0027os_locking_ok\u0027.\n\nChange-Id: Iba7013dd6e1b1e4650b25cd4dd8dc1f355ceb538\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/8685ed4a4cd3c886a00889bb150fbb3ba2620c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/8685ed4a4cd3c886a00889bb150fbb3ba2620c8d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
