)]}'
{"id":"openstack%2Fpuppet-barbican~780845","triplet_id":"openstack%2Fpuppet-barbican~master~I52e08326079619049e2d436cc468829f8721257f","project":"openstack/puppet-barbican","branch":"master","hashtags":[],"change_id":"I52e08326079619049e2d436cc468829f8721257f","subject":"Accept a list value for p11_crypto_plugin_token_labels","status":"MERGED","created":"2021-03-16 13:29:06.000000000","updated":"2021-03-26 17:29:55.000000000","submitted":"2021-03-26 17:29:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780845","meta_rev_id":"6e052b0700043fa073258d9e87fe817883e454c2","_number":780845,"virtual_id_number":780845,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 17:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-03-26 04:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-03-26 13:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2021-03-26 13:16:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 15:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 04:29:44.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-03-26 08:53:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-03-26 08:53:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-26 08:53:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a3f8bc0f5047c29bca30241d4e20c56221e31b20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 13:29:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7763f8cd33948f82789491c3394140ca01bb96e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 13:30:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0aa39a3fd8030b8844c26b6c30ee0373c0878b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 15:17:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/511b3a0b6e3049748e6c565149615e0c : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3eeb8da3f3164fcdad25d61f7610449d : SUCCESS in 5m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/de36bc5f30a6418ba5691cfc6044d0ef : FAILURE in 10m 41s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca40a861844143d88123c0cfab0ad954 : FAILURE in 11m 46s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed761545e04d44acb3be410cb4da31a1 : FAILURE in 9m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54decc372a4240489c2e1eff026bc998 : FAILURE in 9m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46a38b23a2d344b698b59a28e890e561 : SUCCESS in 54m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eb261a40eb6343b182c948fcc4758081 : SUCCESS in 1h 39m 46s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb4cf1d7b36d428da3ac1b6cc07ad74f : SUCCESS in 1h 27m 30s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/31055f5a526a42d3959eda0fdda35c72 : SUCCESS in 18m 03s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/12afa3f16ba54a468314cf6b3503da1b : SUCCESS in 17m 05s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6681182ae2284094bbf859603611fea5 : SUCCESS in 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7faca9ec76625a2a34fecbf88dddc7ac2d4c7c0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-16 22:38:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a0b5c008fa9de5948a3360026eee79ec064b4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 00:20:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a2e15a31d334d0593435e57b1c3afed : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e85e21d4a6cd419c9826e29689de2c1d : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/54fcf9d893c74f118647aae895cf87cf : FAILURE in 8m 48s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/9707fa31de2743308ce9cc9ad280a14f : FAILURE in 8m 24s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e97b020975d14767939ec9979c9c8723 : FAILURE in 7m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7608c07d40434fdc9f82aa2a40f689e4 : FAILURE in 6m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dd2cec857f74575ae04e251257ffe82 : SUCCESS in 47m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/93fe7962bf754a8f996467c550aa9732 : SUCCESS in 1h 30m 36s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/490a07f621fa4cabb696c12f4950da34 : SUCCESS in 1h 22m 52s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b0397268da314d2d9a5e113a45eba485 : SUCCESS in 15m 05s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b191faf7a9847bc811157ed075870f7 : SUCCESS in 15m 19s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/203eea367ce44c37b3e5157c846fa443 : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9bc8117649a9f2e6b5d2d2c6277bde6fe735c9b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-17 08:33:52.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8fdeef38a44650ee3a3dc4f570ecf037913b9ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 10:41:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b07f829d8f2b48699656b888ee1d108f : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce9fbba65dde4ee1b9a50161a70073c8 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9eb0a73680b74e73b21432cd23f6bbc7 : FAILURE in 9m 43s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e17d89116274d62b2a6dc824ad4ced9 : FAILURE in 10m 01s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a1bebb6716a4240a6d4a0105e080b59 : FAILURE in 7m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da8a6c2035d649479da331e2c39d2773 : FAILURE in 7m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4aa060d2584e4105a6c6c4b3bf343845 : SUCCESS in 50m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c154ce39cbb74f8b8882edf1c4e58138 : TIMED_OUT in 2h 00m 48s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f77f18f6a4c42f59bdee48c1ea1ad10 : SUCCESS in 1h 26m 19s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/fe5935fa264e462394a6733347615f8e : SUCCESS in 17m 09s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ccf126eb5c648d6a44840f370c569ad : SUCCESS in 18m 33s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02185ad8e5014d77891ff70c3d6ee866 : SUCCESS in 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"77a0bdfc6c3048d52bb0308c43b3eabd21c6f0f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-20 12:19:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8c795e661a935f24ef8f377857fbbcc435d108d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 13:43:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40bb8d0c6bea40429a3e1e6d81c84060 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48943a8cd6564b72b424e04ef792e35b : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/921af055e78c4f2e937ca33ec7bf08fe : FAILURE in 8m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f74df8508ac3482d866e26be781d0811 : FAILURE in 5m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3a8468276cf4cb59fe378870875086b : FAILURE in 6m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd596c8dc92f46de839edf912fd5fc5e : SUCCESS in 46m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cff9eee30f0b4648bdb0e8cc6bbe8cb8 : SUCCESS in 1h 21m 52s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c1999ece9ae4850aeafad09263ea46a : SUCCESS in 15m 16s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/578ba3762cf54726a26483a105806f5b : SUCCESS in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"262ba3e1a6fbf55c37c21e88a619dddf4c8ea99a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-25 23:38:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"86dcbf44f467c7dd95518a80cf6e3383a756ec6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 00:40:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6b9006b91d14fc8822a9a39845ac27b : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0832eb871f4e039ccbef773aa58fa9 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/80e22beb1e71462f82c6640fcec65615 : SUCCESS in 9m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf392d68fb04cc58eb96d5b274220e7 : SUCCESS in 7m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3078ad9489dd43f48b1c3688ea19d95f : SUCCESS in 7m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb046a16034a42c98432a566223c2bd5 : SUCCESS in 49m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39de8ca054e840ff94a06874e93fc207 : SUCCESS in 59m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/40ba6a5bbd194c81b6f28aab72e53598 : SUCCESS in 16m 35s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e429d0c5ab6a468e97b4bc266ad326f7 : SUCCESS in 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7742f7652f902269abd96e450b54dbed72476775","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-26 04:29:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d00cef6a3ac8c2e4d496f4cf1fe06abc8d0f55fc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-26 13:16:34.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2f66cbe9330a4f6992950caa41e135577cd44c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 13:17:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"173661a4d36d35e80769bc3708123959aa544ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:33:22.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e9f5ea03b01448cb608ed43dc43a3ca : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e8622922a5f41e3b32ed3536c25269a : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/364cee2724bf47358be5c3de5fa6f49f : SUCCESS in 9m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcb26c7f9e8c4a4183dc468e4f166118 : SUCCESS in 8m 04s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b8325ecaa2b4c4388c9c15e47f1b63c : FAILURE in 1h 12m 30s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/295041986b05484f995b49b0db03c38f : SUCCESS in 16m 18s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3962d81dddc74efa8b4227fff1b4218e : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fe230271357b62eff5c7ad4e7d96c98f0546c972","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 14:48:54.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"235095d787c9bb6dcd43a41d32246347bb0bd0ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 16:13:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29038ac8a3424e70a6a2e9a3ab1754e2 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/271e742f686b4dcdb72c672e3d568eea : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1137ff806b549c5b351c5cb824f78ca : SUCCESS in 9m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4f922fedb104e918f0055feccab1481 : SUCCESS in 7m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b145653e071542b893c4af881e81f760 : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9763671677d54d7eb3ba6696a6d24f80 : SUCCESS in 53m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f634320a3eb4d908a235f3d787a7fc4 : SUCCESS in 1h 22m 23s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/af7effd9c2e84e9fbdc44f04a2fe8618 : SUCCESS in 19m 34s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9ff29e138f9402da136db86d6e4f7fe : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3c8d11abbbe98bd59ce99e05793423014fade2c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 16:14:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e74bd3b166ae87be580b74c07d705f30b7dcad43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 17:29:54.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfbc265df8264f278dfb0cf0fa13616a : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ab2821e501e4f0d82a0500b0dc48d57 : SUCCESS in 4m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8a1a871956149f0aa8eecd21ca2d0b1 : SUCCESS in 9m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5d3a9833b3f4350835e27417ca10f32 : SUCCESS in 9m 47s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/40c4f4c1c0a547d286c6e121dc7fc3ce : SUCCESS in 1h 10m 38s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/64baa7457fb841d2afbdad3f1515be72 : SUCCESS in 16m 56s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f7ad45c272a4abe872ead7a82e9949e : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6e052b0700043fa073258d9e87fe817883e454c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 17:29:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b4e7b594ed95227b1040efd1848451f135f873fe","revisions":{"e8a356df685bf1accb08e56e34b9033eb567dc09":{"kind":"REWORK","_number":1,"created":"2021-03-16 13:29:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/780845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/45/780845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/1"}}},"commit":{"parents":[{"commit":"8685ed4a4cd3c886a00889bb150fbb3ba2620c8d","subject":"Add new options to PKCS#11 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/8685ed4a4cd3c886a00889bb150fbb3ba2620c8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:27:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:27:28.000000000","tz":540},"subject":"Allow passing a list to p11_crypto_plugin_token_labels","message":"Allow passing a list to p11_crypto_plugin_token_labels\n\nThe p11_crypto_plugin_token_labels parameter is supposed to take a list\nvalue thus it would be useful if a list value is accepted at puppet\nlayer as well.\n\nChange-Id: I52e08326079619049e2d436cc468829f8721257f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/e8a356df685bf1accb08e56e34b9033eb567dc09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/e8a356df685bf1accb08e56e34b9033eb567dc09"}]},"branch":"refs/heads/master"},"225e61cf5dffdf3dc52c913be25537d91c606d74":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-16 13:30:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/780845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/45/780845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/2"}}},"commit":{"parents":[{"commit":"8685ed4a4cd3c886a00889bb150fbb3ba2620c8d","subject":"Add new options to PKCS#11 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/8685ed4a4cd3c886a00889bb150fbb3ba2620c8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:27:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:29:33.000000000","tz":540},"subject":"Accept a list value for p11_crypto_plugin_token_labels","message":"Accept a list value for p11_crypto_plugin_token_labels\n\nThe p11_crypto_plugin_token_labels parameter takes a list value, thus\nit would be useful if a list value is accepted at puppet layer as well.\n\nChange-Id: I52e08326079619049e2d436cc468829f8721257f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/225e61cf5dffdf3dc52c913be25537d91c606d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/225e61cf5dffdf3dc52c913be25537d91c606d74"}]},"branch":"refs/heads/master"},"ee338e2c0d21e691730592061241254da0cef2f4":{"kind":"REWORK","_number":3,"created":"2021-03-16 22:38:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/780845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/45/780845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/3"}}},"commit":{"parents":[{"commit":"4300d574f5c03c8fcb92b049ace2ebed7a5cd441","subject":"Deprecate support for [p11_crypto_plugin]token_label","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4300d574f5c03c8fcb92b049ace2ebed7a5cd441"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:27:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 22:38:04.000000000","tz":540},"subject":"Accept a list value for p11_crypto_plugin_token_labels","message":"Accept a list value for p11_crypto_plugin_token_labels\n\nThe p11_crypto_plugin_token_labels parameter takes a list value, thus\nit would be useful if a list value is accepted at puppet layer as well.\n\nChange-Id: I52e08326079619049e2d436cc468829f8721257f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ee338e2c0d21e691730592061241254da0cef2f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ee338e2c0d21e691730592061241254da0cef2f4"}]},"branch":"refs/heads/master"},"f7e75efeadf88da365519a6f833e712931d1c10b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-17 08:33:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/780845/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/45/780845/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/4"}}},"commit":{"parents":[{"commit":"4403fe7247151d324f80b865cc2d5a0a078d25fa","subject":"Deprecate support for [p11_crypto_plugin]token_label","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4403fe7247151d324f80b865cc2d5a0a078d25fa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:27:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-17 08:33:42.000000000","tz":540},"subject":"Accept a list value for p11_crypto_plugin_token_labels","message":"Accept a list value for p11_crypto_plugin_token_labels\n\nThe p11_crypto_plugin_token_labels parameter takes a list value, thus\nit would be useful if a list value is accepted at puppet layer as well.\n\nChange-Id: I52e08326079619049e2d436cc468829f8721257f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/f7e75efeadf88da365519a6f833e712931d1c10b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/f7e75efeadf88da365519a6f833e712931d1c10b"}]},"branch":"refs/heads/master"},"0166e144447825f689fb8c3339550f50be2b001c":{"kind":"REWORK","_number":5,"created":"2021-03-20 12:19:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/780845/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/45/780845/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/5"}}},"commit":{"parents":[{"commit":"a89b6a97ee9ccef0e4d30aadb9f6bdae0cf240be","subject":"Merge \"Deprecate support for [p11_crypto_plugin]token_label\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a89b6a97ee9ccef0e4d30aadb9f6bdae0cf240be"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:27:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-20 12:19:29.000000000","tz":540},"subject":"Accept a list value for p11_crypto_plugin_token_labels","message":"Accept a list value for p11_crypto_plugin_token_labels\n\nThe p11_crypto_plugin_token_labels parameter takes a list value, thus\nit would be useful if a list value is accepted at puppet layer as well.\n\nChange-Id: I52e08326079619049e2d436cc468829f8721257f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0166e144447825f689fb8c3339550f50be2b001c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0166e144447825f689fb8c3339550f50be2b001c"}]},"branch":"refs/heads/master"},"b4e7b594ed95227b1040efd1848451f135f873fe":{"kind":"REWORK","_number":6,"created":"2021-03-25 23:38:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/780845/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/45/780845/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/45/780845/6"}}},"commit":{"parents":[{"commit":"f2226dcb12b886cc3a6c0752dbe5625e79dbbf5a","subject":"Add support for oslo_policy/enforce_new_defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/f2226dcb12b886cc3a6c0752dbe5625e79dbbf5a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-16 13:27:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-25 23:38:36.000000000","tz":540},"subject":"Accept a list value for p11_crypto_plugin_token_labels","message":"Accept a list value for p11_crypto_plugin_token_labels\n\nThe p11_crypto_plugin_token_labels parameter takes a list value, thus\nit would be useful if a list value is accepted at puppet layer as well.\n\nChange-Id: I52e08326079619049e2d436cc468829f8721257f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/b4e7b594ed95227b1040efd1848451f135f873fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/b4e7b594ed95227b1040efd1848451f135f873fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
