)]}'
{"id":"openstack%2Fpuppet-barbican~798769","triplet_id":"openstack%2Fpuppet-barbican~stable%2Fvictoria~Ica2918b3235f22348598507ec84b264e627322db","project":"openstack/puppet-barbican","branch":"stable/victoria","topic":"at-params-stable/victoria","hashtags":[],"change_id":"Ica2918b3235f22348598507ec84b264e627322db","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2021-06-30 09:21:52.000000000","updated":"2021-06-30 13:15:26.000000000","submitted":"2021-06-30 13:14:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798769-at-params-stable/victoria","cherry_pick_of_change":757415,"cherry_pick_of_patch_set":1,"meta_rev_id":"36c72978f5d1792196fae868b17115906482dbe0","_number":798769,"virtual_id_number":798769,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 13:14:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-30 11:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-30 09:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-30 11:31:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 09:21:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-30 11:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0abea2aac408beadbeb13d5d003e676021943f40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 09:21:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"16398bf791f01c6594e02cd3859181ba4b3f3b5a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 09:22:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"548ba1cd6bf42e0da575529c7d29bdb8d2ed053c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 11:15:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/545555a086674705ad6b4e6fbec6dfad : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57a6b487669e4a06a9d2c69b528425b1 : SUCCESS in 6m 05s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/c77dabda37a74cd68d26cb88e3ebe91b : SUCCESS in 13m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e0f7d7f94f94daeafb4af1960871703 : SUCCESS in 11m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30eb9b1f1e52400ab625d4187cf29e91 : SUCCESS in 9m 46s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/622839ef165e421ea5674a81f96498c1 : SUCCESS in 57m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42e92163a300436580f7b371af33eeb7 : SUCCESS in 1h 47m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0330c98e41604eb0b24b7b19b669b2fe : SUCCESS in 1h 27m 16s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/58a8309c383a4545a4297c80fbe1b6ca : SUCCESS in 20m 59s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/06fd8e2dcf6441869004f6036aebdf03 : SUCCESS in 25m 37s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c2d49542a6c40358334589d999ea486 : SUCCESS in 15m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdc9ab826ac44a4bb7eebe8cd0fdbfeb : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ccad57628b2556dbabd638f327f5e182cf5932ed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-30 11:31:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62b1f7077b4556ade22db1bbdbfc297f1bc65e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 11:32:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e117574086ba71fddc94edcfe61ec3e5f1301c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:14:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd69933f2d9543b2b84666bb05f6aaea : SUCCESS in 8m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67d28877167e4c02bc5941f233387280 : SUCCESS in 6m 18s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/6971c97637974ea9ac9a2468986b02d5 : SUCCESS in 14m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a885c566db44abda1130ed78acdb85b : SUCCESS in 11m 30s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e789dd0669440819f9a99bcfee2d28e : SUCCESS in 58m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3ad16cfbfcb0419fa5756332700c7a9e : SUCCESS in 1h 38m 41s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1447a13029324156a37b1509c053f859 : SUCCESS in 24m 32s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdc874d811de4b7e9258270b8e235020 : FAILURE in 20m 33s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c757d6e28214eb88b76d92fc3c72da0 : SUCCESS in 17m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ce15ca525d849e7a0e94de8597a0e99 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f9d1c85820d9fdaf8f1666046c3e0cac9e73fad4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:14:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36c72978f5d1792196fae868b17115906482dbe0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:15:26.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8056091417154e5f8141428aa6c71539 : FAILURE in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44717db550ed9ccf1d0f080fc76be991a5b720b2","revisions":{"44717db550ed9ccf1d0f080fc76be991a5b720b2":{"kind":"REWORK","_number":1,"created":"2021-06-30 09:21:52.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/69/798769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/69/798769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/69/798769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/69/798769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/69/798769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/69/798769/1"}}},"commit":{"parents":[{"commit":"5d7568ac1e177b463713243fb224e41bbda462d1","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/5d7568ac1e177b463713243fb224e41bbda462d1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-12 05:02:49.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-06-30 09:21:52.000000000","tz":0},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: Ica2918b3235f22348598507ec84b264e627322db\n(cherry picked from commit bded1a8c1e6c0d2db30da2e1b0f04e1f04cb493f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/44717db550ed9ccf1d0f080fc76be991a5b720b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/44717db550ed9ccf1d0f080fc76be991a5b720b2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
