)]}'
{"id":"openstack%2Fpuppet-barbican~806147","triplet_id":"openstack%2Fpuppet-barbican~master~I60b68cb78f63320fd1df46e0dccda2e8babe28ba","project":"openstack/puppet-barbican","branch":"master","topic":"authtoken-ut","hashtags":[],"change_id":"I60b68cb78f63320fd1df46e0dccda2e8babe28ba","subject":"Do not test authtoken parameters directly","status":"MERGED","created":"2021-08-26 11:50:06.000000000","updated":"2021-09-03 07:06:32.000000000","submitted":"2021-09-03 07:06:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"806147-authtoken-ut","meta_rev_id":"766b8a7ca50f36df700151413f94b5a165c49a0e","_number":806147,"virtual_id_number":806147,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 07:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-03 06:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-09-02 15:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-09-03 06:49:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 12:35:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 13:47:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-09-02 15:53:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-03 06:49:29.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"bdfe7705108780795683a430323f9ae1aaf6bfb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 11:50:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69968ddc240cec7d316dd32028b16b402b34d64a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 11:50:41.000000000","message":"Topic system-scope removed","accounts_in_message":[],"_revision_number":1},{"id":"de737655a14bc4fdce14f7fdf100d4a6999a49ec","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 11:50:43.000000000","message":"Topic set to authtoken-ut","accounts_in_message":[],"_revision_number":1},{"id":"930891864c708b6ce80b8ae9038c086da8e5ea5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 11:54:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b66f296cbb12226d7f2610832450384c20378f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:05:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81a0101e18aa72d4acf73114759ca0e22e49ad18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:14:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c7f16071ef9dd6074279a316db41ecb24b17319c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:14:40.000000000","message":"Topic system-scope removed","accounts_in_message":[],"_revision_number":3},{"id":"e80f77d55c17006944a9e690831cd66fd71eb42d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:14:48.000000000","message":"Topic set to authtoken-ut","accounts_in_message":[],"_revision_number":3},{"id":"5cd14abf81967452f5e0a7f2721f6e46bf9f55da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:35:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b6a18dc0f7941c182c6241d1b8ed523 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/732da9288b7644dab654c9d7421fc9a3 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/220bbc43a7fe4261851925a8dbb6e1ad : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9e5a13c6500404892141cd3b4266276 : SUCCESS in 12m 07s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f873a3aa71a1409084a74ede63e78f71 : SUCCESS in 13m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9da372511abf4e2ab626c8742c8cbef8 : SUCCESS in 9m 05s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13c034a8fbda4344a3fcfef2f3be08a0 : SUCCESS in 9m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdbadea2f37946fea820f55d4c22a80c : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68f79386f5c9fd4aac8b0ebdeb00c68ea658960b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:42:20.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"2af58fb7f13f605414dd0a006cadc692911e9e05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:55:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"90e0dae813b49df2a16fbbfc2d5abd6aee0103d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 13:09:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1f07b1c2d6e5226e4df9d1b2a8a5bc4fddacfd8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 13:26:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8988c7d311145bdbe97b9a1b9e0d06b : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f367f8820a0b4999aeb221a053f16e46 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50978c6c0103495a81430442aed0e82f : SUCCESS in 5m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/788220a39db04a518214948f976e6a32 : SUCCESS in 14m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d0511ca1b664a8094cd423448a354e2 : SUCCESS in 13m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccb63f09d5664bf28c1e22f782c833de : SUCCESS in 11m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4571f3a663764079ac531c48e6f8a144 : SUCCESS in 9m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b31ac67ced9f44d18115e84aa88e886c : SUCCESS in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1d63b460b7582dd39fa8d00cecfe7d4e1cd3c639","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-26 13:47:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9c2da249280528c8034565301908c2b1fa7090ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 21:23:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bcbb1d5dab2a8f8a53ff205cea9a077abc6ffd1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 21:44:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c11721536b9047dea26ca229129d304d : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e36f09d457e94cc796398156fd6f6c4c : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7f9d56d57c24ab3ac3b31b48a440f8b : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a276ef405ad4cd8b7c67448276e5afc : SUCCESS in 12m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a26d84793294405f9a353bb3d1690a59 : SUCCESS in 11m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6564503d125c42b189324a28d94d99b4 : SUCCESS in 9m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bfa9fb42a554747ab6fd197fcbadc0c : SUCCESS in 12m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d37a8ba4fb974245aafad3a3399ccd4e : SUCCESS in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"207c0bcac6b86f949629c1318e62ec52cf75c23d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 15:53:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5f8674883d28d3168187f7ce743894df0d0b4e05","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-03 06:49:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2ce9c1ae6b1f08cdb40b8460e01d429d685c5e1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:50:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f8873ebca69fa0b6eefc24f0738410baf3fa7ba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:06:31.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49767c04f45f4a328c33050e47357931 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/770755eaa282426e84e41489c160e85e : SUCCESS in 4m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24d065bf68bc42a3b29665a029840f22 : SUCCESS in 4m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4cddad14d27416e9876ff6fc5acfd09 : SUCCESS in 13m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cbe8bb42bd741cf9d7904a48b3bd554 : SUCCESS in 14m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb7727ff731642a5b06d8c0e366ef659 : SUCCESS in 14m 56s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26ed350ee34143119b83d404a7c0785a : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":6},{"id":"766b8a7ca50f36df700151413f94b5a165c49a0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:06:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4fbb82c79aa1366a5f44841de4e47eab0a8b97c6","revisions":{"fb40d968e1ae6ca1ecd6426a7e233a22efa15197":{"kind":"REWORK","_number":1,"created":"2021-08-26 11:50:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/806147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/47/806147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/1"}}},"commit":{"parents":[{"commit":"90f816303f672a2b014b6dcd4b9f5ee9385219ba","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/90f816303f672a2b014b6dcd4b9f5ee9385219ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:49:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:49:38.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I60b68cb78f63320fd1df46e0dccda2e8babe28ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/fb40d968e1ae6ca1ecd6426a7e233a22efa15197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/fb40d968e1ae6ca1ecd6426a7e233a22efa15197"}]},"branch":"refs/heads/master"},"dc59a6d8bfe7eb96577dd90459484c9ef361aa89":{"kind":"REWORK","_number":2,"created":"2021-08-26 12:05:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/806147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/47/806147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/2"}}},"commit":{"parents":[{"commit":"90f816303f672a2b014b6dcd4b9f5ee9385219ba","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/90f816303f672a2b014b6dcd4b9f5ee9385219ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:49:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:05:40.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I60b68cb78f63320fd1df46e0dccda2e8babe28ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/dc59a6d8bfe7eb96577dd90459484c9ef361aa89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/dc59a6d8bfe7eb96577dd90459484c9ef361aa89"}]},"branch":"refs/heads/master"},"520fd0d487c77b6da592d84355af482c75419b94":{"kind":"REWORK","_number":3,"created":"2021-08-26 12:14:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/806147/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/47/806147/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/3"}}},"commit":{"parents":[{"commit":"90f816303f672a2b014b6dcd4b9f5ee9385219ba","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/90f816303f672a2b014b6dcd4b9f5ee9385219ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:49:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:13:53.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I60b68cb78f63320fd1df46e0dccda2e8babe28ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/520fd0d487c77b6da592d84355af482c75419b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/520fd0d487c77b6da592d84355af482c75419b94"}]},"branch":"refs/heads/master"},"50bc9bad3b834cce3a24cf0daa310bfdfe2f3f46":{"kind":"REWORK","_number":4,"created":"2021-08-26 12:55:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/806147/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/47/806147/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/4"}}},"commit":{"parents":[{"commit":"90f816303f672a2b014b6dcd4b9f5ee9385219ba","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/90f816303f672a2b014b6dcd4b9f5ee9385219ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:49:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:55:31.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I60b68cb78f63320fd1df46e0dccda2e8babe28ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/50bc9bad3b834cce3a24cf0daa310bfdfe2f3f46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/50bc9bad3b834cce3a24cf0daa310bfdfe2f3f46"}]},"branch":"refs/heads/master"},"44c69610c74302bd8249c29115cda72d15e358f5":{"kind":"REWORK","_number":5,"created":"2021-08-26 13:09:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/806147/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/47/806147/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/5"}}},"commit":{"parents":[{"commit":"90f816303f672a2b014b6dcd4b9f5ee9385219ba","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/90f816303f672a2b014b6dcd4b9f5ee9385219ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:49:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:09:45.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I60b68cb78f63320fd1df46e0dccda2e8babe28ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/44c69610c74302bd8249c29115cda72d15e358f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/44c69610c74302bd8249c29115cda72d15e358f5"}]},"branch":"refs/heads/master"},"4fbb82c79aa1366a5f44841de4e47eab0a8b97c6":{"kind":"REWORK","_number":6,"created":"2021-08-26 21:23:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/806147/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/47/806147/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/47/806147/6"}}},"commit":{"parents":[{"commit":"90f816303f672a2b014b6dcd4b9f5ee9385219ba","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/90f816303f672a2b014b6dcd4b9f5ee9385219ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 11:49:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 21:23:15.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I60b68cb78f63320fd1df46e0dccda2e8babe28ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4fbb82c79aa1366a5f44841de4e47eab0a8b97c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4fbb82c79aa1366a5f44841de4e47eab0a8b97c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
