)]}'
{"id":"openstack%2Fpuppet-barbican~807430","triplet_id":"openstack%2Fpuppet-barbican~master~I8e2d912f75bb36b129a25ddf9d6323d1adadd49b","project":"openstack/puppet-barbican","branch":"master","topic":"remove_empty_pol_file","hashtags":[],"change_id":"I8e2d912f75bb36b129a25ddf9d6323d1adadd49b","subject":"Allow purging policy files","status":"MERGED","created":"2021-09-04 12:46:05.000000000","updated":"2021-09-20 08:11:48.000000000","submitted":"2021-09-20 08:09:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807430-remove_empty_pol_file","meta_rev_id":"bc76e538cd8b71ec69f7c610a6d5ad5501d3d59d","_number":807430,"virtual_id_number":807430,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 08:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-09-09 16:02:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-09-20 06:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-09-20 06:02:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-04 14:29:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 16:02:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-20 06:02:12.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"578b01a02e54ba6bb0bbe6b4cb29e332b8e669fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 12:46:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f1061cbc758e229a54ad56ffa3bb8c5d5909cd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 12:56:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"71b8485aad4ebf4929fb815d16ac87a0739c1042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 14:29:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/629b2b01ad0140a09409ee4ec2a33e54 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41f76773cced4646b02692fa43017307 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e78353ff1de4244b465295482c13cc1 : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f975be2a1054e008b5f40fa083374c2 : SUCCESS in 13m 52s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/50fa9d424e884323842ee1d0773ca064 : SUCCESS in 9m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3f9f9a4749e4d1c887a455889db8af9 : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bcd7636ff5845d0978b38ae5e07c6c9 : SUCCESS in 9m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6f15b7ce8b04018bc7fd6bfd158f193 : SUCCESS in 9m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9ccc4f3960347e5b6a4517c2ccc787d : SUCCESS in 49m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a899aade07e3450b9a487a939dc5b8d8 : SUCCESS in 1h 27m 36s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc7469ac3c2e4fdc957fa799217a1b9c : SUCCESS in 17m 02s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef2ec1636c8740b2a4395b9fbe888307 : SUCCESS in 25m 02s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/247f205aba5343969b7e127179fad938 : SUCCESS in 14m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4e4083f7b7d45b28542e08129af976f : SUCCESS in 17m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d8bfa40ca134a1fbb59d62e18bc900f : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"589ba21cdb9ba8e50796e4c00d6c6aa9413496af","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-09 16:02:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9301f49fe8c8d8d1bff911d5f6f57016cb73ecb4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-09-20 06:02:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"545c914f21a4e9aaf43afe1b4ea51a6375e363d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 06:10:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8522982d8e1e3f84327555a481506bc7ef60a860","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:09:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27d991edcaa54586895725a4dbf7f53b : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa6e7b302de14a6fa6522926c7945cb1 : SUCCESS in 4m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7bc8450b45f4fa7915918b439b9c02a : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/39e3fa212c3c49679d4eb3db9f3584f2 : SUCCESS in 12m 26s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cf338ab9eb240ad9a650392bad01982 : SUCCESS in 13m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6445734b551d49e594b52e9c441a986e : SUCCESS in 9m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3733f09c75840d5a0f4c4d9d5aac06c : SUCCESS in 11m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fd0cefde6a84ba9a380c53008566206 : SUCCESS in 1h 33m 50s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5faf811c312a485da3396923a3cb43be : SUCCESS in 22m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3398ab3e66e246428a043f91902b0b1b : SUCCESS in 14m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2a0ca0d89454b5394ae1cfb7a75b27a : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b4a7d172f0de507b29ecc2f421056ef5845e4872","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:09:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bc76e538cd8b71ec69f7c610a6d5ad5501d3d59d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:11:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28ec4bc474414e5c9602e1e05d3f0c97 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46265f57c5776d4eb90496ec3cb290591818f7c1","revisions":{"0f445a6bf708a62c7ddb1998618a06cff39cdcdd":{"kind":"REWORK","_number":1,"created":"2021-09-04 12:46:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/807430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/30/807430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/1"}}},"commit":{"parents":[{"commit":"4fbb82c79aa1366a5f44841de4e47eab0a8b97c6","subject":"Do not test authtoken parameters directly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4fbb82c79aa1366a5f44841de4e47eab0a8b97c6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 12:45:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 12:45:54.000000000","tz":540},"subject":"Delete policy file if policies hash is empty","message":"Delete policy file if policies hash is empty\n\nIf all policies get deleted, previously created policy files won\u0027t\nget removed and the api policy file is still used. Let\u0027s make\nsure the policy file gets purged if the policies hash is empty.\n\nCo-Authored-By: Martin Schuppert \u003cmschuppert@redhat.com\u003e\nDepends-On: https://review.opendev.org/802305\nChange-Id: I8e2d912f75bb36b129a25ddf9d6323d1adadd49b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0f445a6bf708a62c7ddb1998618a06cff39cdcdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0f445a6bf708a62c7ddb1998618a06cff39cdcdd"}]},"branch":"refs/heads/master"},"46265f57c5776d4eb90496ec3cb290591818f7c1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-04 12:56:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/807430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/30/807430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/30/807430/2"}}},"commit":{"parents":[{"commit":"4fbb82c79aa1366a5f44841de4e47eab0a8b97c6","subject":"Do not test authtoken parameters directly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/4fbb82c79aa1366a5f44841de4e47eab0a8b97c6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 12:45:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 12:55:59.000000000","tz":540},"subject":"Allow purging policy files","message":"Allow purging policy files\n\nThis change introduces the new purge_config parameter to the policy\nclass so that any policy rules not managed by puppet manifests can be\ncleared.\n\nCo-Authored-By: Martin Schuppert \u003cmschuppert@redhat.com\u003e\nDepends-On: https://review.opendev.org/802305\nChange-Id: I8e2d912f75bb36b129a25ddf9d6323d1adadd49b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/46265f57c5776d4eb90496ec3cb290591818f7c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/46265f57c5776d4eb90496ec3cb290591818f7c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
