)]}'
{"id":"openstack%2Fpuppet-barbican~813017","triplet_id":"openstack%2Fpuppet-barbican~master~I989b99f2ba68dd2782bd43b91b002ca290b48656","project":"openstack/puppet-barbican","branch":"master","hashtags":[],"change_id":"I989b99f2ba68dd2782bd43b91b002ca290b48656","subject":"Rename redundant parameters of barbican::keystone::notification","status":"ABANDONED","created":"2021-10-07 14:09:52.000000000","updated":"2022-05-23 00:16:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"22884330170d33356f05170fc8bf9473c75fd03b","_number":813017,"virtual_id_number":813017,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-07 16:49:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-10-07 21:45:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2021-10-07 23:10:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 14:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 21:45:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c41518b2447d6c96a3475731c3235786532a41e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 14:09:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ddd81569d0eafa07a0e12a9319574622527da8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 14:34:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a65aff738e1245f198be9c4791e66681 : FAILURE in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ded9be4249b4a5788ab5ae86c7c0e18 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95f778704bf04c0b851aa049d9df80ed : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2f1c3b6c905484d9f22a30020dd4153 : FAILURE in 15m 58s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/626ca3b374e149209b57d9133a6792cc : FAILURE in 18m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03f1a98e23b8402ab7a1e36b718ffe72 : FAILURE in 9m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f744169c9134c359a8d7927261860d7 : FAILURE in 9m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b8170ab9575454593a2b1c462ba0c5e : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c848d093236542a5b593cedb1c973a90 : FAILURE in 8m 17s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/460b332571904fa8a2c01238f5042143 : FAILURE in 13m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547a16ad619f4002989cb82cf95d6397 : FAILURE in 7m 30s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce3a409a10f143f981d39000603e01cc : FAILURE in 11m 17s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bfa17d0dd9b4ad7b3864ed94ffbd337 : FAILURE in 10m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6926158295db4c9686fb508664177646 : FAILURE in 8m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dae80d0a81e3497a858a0023077708cf : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1881178d3189d96d62a903fd6b075200355a55a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 15:25:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6601b409cbce0d4241ad5f5219396f58270a5c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:49:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fea9eb74ba947cb8a80f660b3a14537 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff5d3c8982aa47949180c13d782de498 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dbcc10b93a345d9a3ba64cdc74f5bb6 : SUCCESS in 4m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f55f23730a44fce84604cb723818ac9 : SUCCESS in 13m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0718a5cb52aa4062b26292f2fff87593 : SUCCESS in 12m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/073f21f6a5304921ac6cfe4cac77c4f4 : SUCCESS in 12m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1bd02aad99742448433d621cbdb44b5 : SUCCESS in 10m 04s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87f08c108ff441589efc5079f4386510 : SUCCESS in 12m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/910c2f8c6a5747fe91bcd6b0bf3056e9 : FAILURE in 11m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/62f7009f0d55436a8ec2a61042d3b0bc : SUCCESS in 1h 19m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b12c03a0a8347259362d65ec2899d4b : SUCCESS in 18m 36s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad5c162e351c41e1bb58e8e09b8ed231 : SUCCESS in 24m 43s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae3f449fa1f742dfbdc89ef61b6ef494 : SUCCESS in 18m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41601ffe6f7b49ecaf5b5b61b2f4887d : SUCCESS in 17m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36634899a3b940e681f5a41d6c86974a : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1ba38a5e0c1f8102e5ff8868f9f1928b6a748884","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-07 21:45:40.000000000","message":"Patch Set 2: Code-Review+2\n\neh this seems like a rename for rename sake. Not certain it\u0027s better but i\u0027ll defer to the others","accounts_in_message":[],"_revision_number":2},{"id":"66dd707488bf6871ffc037d8e295681c36c69d40","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 23:10:12.000000000","message":"Patch Set 2: Workflow-1\n\nLet me give this another thought. Seems like the class name is also inconsistent with the section name (notification vs notifications) ...","accounts_in_message":[],"_revision_number":2},{"id":"22884330170d33356f05170fc8bf9473c75fd03b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-23 00:16:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bdbd575b4dcab4e3e4d5e7778979f3c89e97817a","revisions":{"a679af6f5998f992788c4c7bb3de13dcdd9fd36c":{"kind":"REWORK","_number":1,"created":"2021-10-07 14:09:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/813017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/17/813017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/1"}}},"commit":{"parents":[{"commit":"deeac557a9ea036715893cb89eb0401db9a3b811","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/deeac557a9ea036715893cb89eb0401db9a3b811"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 14:08:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 14:09:03.000000000","tz":540},"subject":"Rename redundant parameters of barbican::keystone::notification","message":"Rename redundant parameters of barbican::keystone::notification\n\nChange-Id: I989b99f2ba68dd2782bd43b91b002ca290b48656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a679af6f5998f992788c4c7bb3de13dcdd9fd36c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a679af6f5998f992788c4c7bb3de13dcdd9fd36c"}]},"branch":"refs/heads/master"},"bdbd575b4dcab4e3e4d5e7778979f3c89e97817a":{"kind":"REWORK","_number":2,"created":"2021-10-07 15:25:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/813017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/17/813017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/17/813017/2"}}},"commit":{"parents":[{"commit":"deeac557a9ea036715893cb89eb0401db9a3b811","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/deeac557a9ea036715893cb89eb0401db9a3b811"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 14:08:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 15:25:34.000000000","tz":540},"subject":"Rename redundant parameters of barbican::keystone::notification","message":"Rename redundant parameters of barbican::keystone::notification\n\nChange-Id: I989b99f2ba68dd2782bd43b91b002ca290b48656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/bdbd575b4dcab4e3e4d5e7778979f3c89e97817a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/bdbd575b4dcab4e3e4d5e7778979f3c89e97817a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
