)]}'
{"id":"openstack%2Fpuppet-barbican~829283","triplet_id":"openstack%2Fpuppet-barbican~master~I2c8f78924f93c92e6f3af751e41699907db02ea5","project":"openstack/puppet-barbican","branch":"master","topic":"c9-ut","hashtags":[],"change_id":"I2c8f78924f93c92e6f3af751e41699907db02ea5","subject":"DNM: Test unit tests for CentOS 9","status":"ABANDONED","created":"2022-02-15 15:00:14.000000000","updated":"2022-02-15 16:36:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34a2995fb801756b74720d3296af28ba3a64f99c","_number":829283,"virtual_id_number":829283,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-15 16:21:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 16:21:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abe7351b4ec1cd110dc3a2866f0f9ba50eca90cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98b353bb11ae54c04ee37c9439667f33a1fcb852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:05:16.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"414153d5c7ecd27542ffd33c6ff623837baf2ba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:17:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"56e037ffaf0805fca3ac14a0edb9d6a7db601aa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:21:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a52fc953af949d6aa7af2b58ddf378e : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e962c581a3415bb408e44694548e89 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f743dcc03f54545a376da3ab199784e : SUCCESS in 4m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe9c79b98cde458fb6a2d7022eaf019c : SUCCESS in 16m 39s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/65f82c6bd704456eae6c9c04ef89dfb6 : SUCCESS in 16m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05adfabce9f8460e8904c4fcd748200c : SUCCESS in 13m 37s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc6ff5a5320745d4a574c1026f1a07e9 : SUCCESS in 18m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d6e0abff9f642c49b752c00de893a06 : SUCCESS in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34a2995fb801756b74720d3296af28ba3a64f99c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 16:36:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ec9fad4f8e2b350ce423c7747946bdbb54c77683","revisions":{"7ae0661fbb54ee3bfd7d1d2c9585f419f261ed7b":{"kind":"REWORK","_number":1,"created":"2022-02-15 15:00:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/829283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/83/829283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/1"}}},"commit":{"parents":[{"commit":"d64ec669d9bfb5238cbdc7002eafdac3ee4364d3","subject":"Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d64ec669d9bfb5238cbdc7002eafdac3ee4364d3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:59:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:59:58.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\n\nChange-Id: I2c8f78924f93c92e6f3af751e41699907db02ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7ae0661fbb54ee3bfd7d1d2c9585f419f261ed7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7ae0661fbb54ee3bfd7d1d2c9585f419f261ed7b"}]},"branch":"refs/heads/master"},"f15a3ade1a2e336a46e4c24818498d022a7e3378":{"kind":"NO_CHANGE","_number":2,"created":"2022-02-15 15:05:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/829283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/83/829283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/2"}}},"commit":{"parents":[{"commit":"ceb7841178adc109ef9343be4a71d39b569567e3","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ceb7841178adc109ef9343be4a71d39b569567e3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:59:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:05:09.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\n\nChange-Id: I2c8f78924f93c92e6f3af751e41699907db02ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/f15a3ade1a2e336a46e4c24818498d022a7e3378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/f15a3ade1a2e336a46e4c24818498d022a7e3378"}]},"branch":"refs/heads/master"},"ec9fad4f8e2b350ce423c7747946bdbb54c77683":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-15 15:17:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/829283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/83/829283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/83/829283/3"}}},"commit":{"parents":[{"commit":"d8067ab5a2d97a1d2431b199da36df04a469c518","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d8067ab5a2d97a1d2431b199da36df04a469c518"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 14:59:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:17:31.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\n\nChange-Id: I2c8f78924f93c92e6f3af751e41699907db02ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ec9fad4f8e2b350ce423c7747946bdbb54c77683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ec9fad4f8e2b350ce423c7747946bdbb54c77683"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
