)]}'
{"id":"openstack%2Fpuppet-barbican~839233","triplet_id":"openstack%2Fpuppet-barbican~master~I0e20b135e8c29f1d27d39d2940d49ea30a1f512c","project":"openstack/puppet-barbican","branch":"master","topic":"base-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-06 02:23:33.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I0e20b135e8c29f1d27d39d2940d49ea30a1f512c","subject":"Add general basic functionality to the base barbican class","status":"MERGED","created":"2022-04-25 16:07:12.000000000","updated":"2022-05-06 04:49:29.000000000","submitted":"2022-05-06 04:48:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839233-base-opts","meta_rev_id":"adefa09273ca8533df21a69616eaef121b9077e4","_number":839233,"virtual_id_number":839233,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 04:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-04-26 07:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-28 20:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-04-28 20:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-25 16:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-26 07:39:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-04-28 20:00:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"88975b73934d6005c2b3a2ba9bf56da8496aca11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-25 16:07:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34cc1b6339038bcda7036b3a0c247f890bfb76ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 16:23:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c14770fee5e445a7a8777428ca95a1cc : FAILURE in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8583040db2d4814a4c647853c909191 : FAILURE in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517a66bcd8a6470b8b5e11825c504099 : FAILURE in 3m 33s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f84c780c7b2c4724950152b2319b486f : FAILURE in 9m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bef6412321e43afb0fdc6469d81f249 : FAILURE in 9m 41s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/218cbae44d6d426bbcce39c255f3a059 : FAILURE in 8m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2092f611ff0e4466944b3afce9c6f780 : FAILURE in 8m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/764c8bda391746c2a23fc07e62ce7cd8 : FAILURE in 7m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d01f83d803b04a88a4d851ebf40c6e0c : FAILURE in 12m 08s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d611ffe64e57485b8ef02e4ec138aed8 : FAILURE in 6m 57s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1add90d9440424eb459937258b4f857 : FAILURE in 9m 48s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/24587cb1852944ef9c76667b17fac515 : FAILURE in 10m 02s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82047addaaa94e79bd5ae233e5f6a6e2 : FAILURE in 6m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab8a1a858d834fdeb4b00a0a03c8ef5c : FAILURE in 8m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bf695a59b2f46b1b5c471c13d122a2f : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b044fdee9c2246915a6dbf4d747fbdc368c7bd4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-25 23:35:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e02ea92af0718760461660a967220a986bf9684","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-25 23:36:10.000000000","message":"Topic set to base-opts","accounts_in_message":[],"_revision_number":2},{"id":"c9cf0472b90cc50bf7030b7bb9f540826c0a1ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 23:54:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/614f93f9a6e7407487838f373f035245 : FAILURE in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26fdab38052a43f5900cd0e56b83fbf3 : FAILURE in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/164c29d35542401dbcb0ddf183478857 : FAILURE in 3m 27s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/39994c7d99dc403084bf7543c87b082e : FAILURE in 11m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/230e530de15244e79eb2fb41b32ed961 : FAILURE in 8m 32s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f64d260a917f4a6b9bab7a2ee11cb350 : FAILURE in 9m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec1c1a4b5a43497d8b76bc5b8bf1564f : FAILURE in 8m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93b0ffd9037d4caf974c46814a44062f : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ded51e24b6f74b72a9d1ea4a4d3654e9 : FAILURE in 12m 00s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73efd56c318a458fb7cf75ce17f9a7eb : FAILURE in 6m 56s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b125829cda54386bb08def51f8bc543 : FAILURE in 10m 22s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c70aaf8eee1141f1b287b9ed1e62b579 : FAILURE in 9m 00s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b176608179d84a5eacaa1b4a2826e7da : FAILURE in 7m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/361cb0ef2e8e41e0b7b3fd8367c41db7 : FAILURE in 7m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f11a5c798e344d9a941fe69e0c7bc2e : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f532eb4d27d1515b3f58a01f353ec5f0db14ccec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-26 00:01:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a143b0f9748874f12e5433272ca6b07da8faeb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 01:04:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a11fb5e87144e58a31c35b145d0062e : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c65ac71ea164358a66cadf9b9b4fc74 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5fe08a01ffe421dacd3266fa0fdb366 : SUCCESS in 2m 59s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0b2390e886b4839a6b8898191352baf : FAILURE in 9m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4229c3f9fe046689f4ba9ff4b9d75db : FAILURE in 7m 45s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8824f74c20174d03b520c410cc13573e : FAILURE in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a36f7ed585a41178359f87f72770ec3 : FAILURE in 7m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66e23ce0a73f47b3a0baba95a0cb3d9c : FAILURE in 35m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/213e25fdffb64c399982f36c232998eb : SUCCESS in 59m 28s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b90629e89b94303ab8797a7360cbef1 : SUCCESS in 12m 36s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f4333da6846498c996a858a55beecd0 : SUCCESS in 21m 04s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f401d572da344a488dfcdcc99a43cf4a : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2e73d865d9e4a4e8418b31eb8ee50c1 : SUCCESS in 15m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff677280cdaf4430acc1dfb58e9d43c2 : SUCCESS in 13m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50821e903c5742e1965e3f1227b90b36 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"2ab40a363c95406b63a434995a9e956048afd3d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-26 01:06:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"17c201b1fc78004cd34edfe5e287b344c824cf05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 02:07:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da3d918b826a4e94a610bbbaf37af92a : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407d59ab357b46e7b6cb9756b700ed29 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca77a4cfbefa43f0b49bb8ff683a24f1 : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/921c0370f8a6406f839248d792db2d62 : SUCCESS in 12m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/280a97689c3c4f9bbdf040b7b9f64e66 : SUCCESS in 8m 59s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fc78836a5c4488ca22e997dac8025c5 : SUCCESS in 7m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ef5f2b5f5044879c2ba9011e9a2b02 : SUCCESS in 7m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f48a82cc19f47118c52e293a4cadb0c : FAILURE in 37m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f814d62f9f014b6f839eb9fab0908b42 : SUCCESS in 59m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ffbcaadcb2c4b09bd8abfd50a715e8e : SUCCESS in 13m 06s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a3295d110864021bfc6017c7bddf3d7 : SUCCESS in 20m 06s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3eea8f72cc3e419d9e83d10215d6257b : FAILURE in 8m 06s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80dfab4116854e1eb23886893ab956dc : SUCCESS in 13m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18790d0922c549e6aface2d8a6e41147 : SUCCESS in 11m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3a57876c60447069f4b8427cc5e4813 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":4},{"id":"cface8011ef09b8ed883ebac0be134244c4628c1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-26 07:39:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fa2c5edd131b292b144a735f361e8e587219fb04","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-28 20:00:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c0896d2a53bb40ca328deb5f0b78f72bc3e50d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:01:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e134d644bd84f45d4a50162e02ad671a55809f81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:24:35.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e630095728d413dad9fcc37905db3e0 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeef601fd5804315b6b71464d80c48c5 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/735f500f956a4c00b604bc8fa334ac70 : SUCCESS in 3m 19s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/21db1651d2ca4645ac127119856aeabf : SUCCESS in 10m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46504e65cb1c46848a0e51b272130fa4 : SUCCESS in 9m 24s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9963851839c4ff88381d1ce3e748f7f : SUCCESS in 8m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bef459b7b7e43c8b019d52fa4830b29 : FAILURE in 12m 24s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/28f3ec0f9267496ead4090e123872cf8 : SUCCESS in 22m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be6cf9146bbb42098cd4a396df59b92f : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":4},{"id":"7452fb31026bee5009df85c6b11644137d2d3f78","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-06 02:23:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5bca84c90321056c3b210af57eeb2e290dd0264f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 03:36:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d141f1a15d7c4935ad14c10d4a6b0c78 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/426c2660a801453ca42c737e38c01241 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdc8c99be87d4563b07743c50b6756a3 : SUCCESS in 4m 09s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ce1d02b1a634f60810e0617959117ef : SUCCESS in 13m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b48ddd093a9a4791bd8dda0913d65b0d : SUCCESS in 11m 18s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/052ee8e77ab043e9929c640bcbef1e9e : SUCCESS in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de2bd79b5ad44014883d9a9a2377a95a : SUCCESS in 8m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e11a5a92f46f4eacaebf234311db45d0 : FAILURE in 36m 47s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e8b256d07c6491aaf696cc20228ccd6 : SUCCESS in 1h 02m 34s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cfc3c341da046d18f73d1bd92dfa2ed : SUCCESS in 13m 18s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/505bbea51ec3443a9b9be562b9dc2730 : SUCCESS in 19m 44s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/88c4b83a9c584af1ad30ec0950d13101 : FAILURE in 12m 13s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1da8667b45d345b492689393fa220599 : SUCCESS in 14m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cf8e848edcb4d489ba9d821db1554da : SUCCESS in 13m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efa0c9b4eab045d0945574c4a45e4be9 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":4},{"id":"4806e2abf39eb9295e77adfdd8b9c18e298e10d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 03:38:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0e2498ec5d631d6ead19d42266d9e13af08618ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:48:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34d8a9ddae514ae8b0ac9f927d4ada40 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b829ef8ab1c5430787a31cdba8cd86c9 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/675d103b31194349b209d95593fe7ad5 : SUCCESS in 3m 21s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d494ae27cb1747c99a45e076eac9e154 : SUCCESS in 11m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86d91a01c2a64770b4c004f62be9d20a : SUCCESS in 8m 30s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da9c8ac61ecf4c9fab08adf4051c4736 : SUCCESS in 8m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29b40b3e4c2c4fc49c069d9c27900b97 : SUCCESS in 1h 04m 55s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c1fc5427c82402186b93a8439cb8d17 : SUCCESS in 20m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1af358f33f894c04b288f1d8b93ba44e : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":4},{"id":"69fbcbe3ae4a0113af3f7591dd67db0d20ea800e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:48:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"adefa09273ca8533df21a69616eaef121b9077e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:49:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41c93bbfa5cf4dddb6a357f64836fe85 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"700b5e4148baf7ea44a12fcedb7bb278f1f2bcc3","revisions":{"5ae42bcd48783c9e2be6fd81b99e6123c2c8d793":{"kind":"REWORK","_number":1,"created":"2022-04-25 16:07:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/839233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/33/839233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/1"}}},"commit":{"parents":[{"commit":"75587399218023c1fba042747fafa89a1e34545f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/75587399218023c1fba042747fafa89a1e34545f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-25 16:05:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-25 16:07:00.000000000","tz":540},"subject":"Add general basic functionality to the base barbican class","message":"Add general basic functionality to the base barbican class\n\nThis is the prep work to migrate some common parameters from the api\nclass to the base class, and implements basic functionality so that\nthe base class provides consistent functionality in all modules.\n\nChange-Id: I0e20b135e8c29f1d27d39d2940d49ea30a1f512c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/5ae42bcd48783c9e2be6fd81b99e6123c2c8d793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/5ae42bcd48783c9e2be6fd81b99e6123c2c8d793"}]},"branch":"refs/heads/master"},"c18f0de332255d49336f16cef49f43cc1adb5bb8":{"kind":"REWORK","_number":2,"created":"2022-04-25 23:35:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/839233/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/33/839233/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/2"}}},"commit":{"parents":[{"commit":"75587399218023c1fba042747fafa89a1e34545f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/75587399218023c1fba042747fafa89a1e34545f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-25 16:05:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-25 23:35:30.000000000","tz":540},"subject":"Add general basic functionality to the base barbican class","message":"Add general basic functionality to the base barbican class\n\nThis is the prep work to migrate some common parameters from the api\nclass to the base class, and implements basic functionality so that\nthe base class provides consistent functionality in all modules.\n\nChange-Id: I0e20b135e8c29f1d27d39d2940d49ea30a1f512c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/c18f0de332255d49336f16cef49f43cc1adb5bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/c18f0de332255d49336f16cef49f43cc1adb5bb8"}]},"branch":"refs/heads/master"},"d30c129d1d12374983866a903e071c7d2532d816":{"kind":"REWORK","_number":3,"created":"2022-04-26 00:01:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/839233/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/33/839233/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/3"}}},"commit":{"parents":[{"commit":"75587399218023c1fba042747fafa89a1e34545f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/75587399218023c1fba042747fafa89a1e34545f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-25 16:05:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-26 00:01:13.000000000","tz":540},"subject":"Add general basic functionality to the base barbican class","message":"Add general basic functionality to the base barbican class\n\nThis is the prep work to migrate some common parameters from the api\nclass to the base class, and implements basic functionality so that\nthe base class provides consistent functionality in all modules.\n\nChange-Id: I0e20b135e8c29f1d27d39d2940d49ea30a1f512c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d30c129d1d12374983866a903e071c7d2532d816"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d30c129d1d12374983866a903e071c7d2532d816"}]},"branch":"refs/heads/master"},"700b5e4148baf7ea44a12fcedb7bb278f1f2bcc3":{"kind":"REWORK","_number":4,"created":"2022-04-26 01:06:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/839233/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/33/839233/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/33/839233/4"}}},"commit":{"parents":[{"commit":"75587399218023c1fba042747fafa89a1e34545f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/75587399218023c1fba042747fafa89a1e34545f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-25 16:05:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-26 01:06:23.000000000","tz":540},"subject":"Add general basic functionality to the base barbican class","message":"Add general basic functionality to the base barbican class\n\nThis is the prep work to migrate some common parameters from the api\nclass to the base class, and implements basic functionality so that\nthe base class provides consistent functionality in all modules.\n\nChange-Id: I0e20b135e8c29f1d27d39d2940d49ea30a1f512c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/700b5e4148baf7ea44a12fcedb7bb278f1f2bcc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/700b5e4148baf7ea44a12fcedb7bb278f1f2bcc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
