)]}'
{"id":"openstack%2Fpuppet-barbican~947591","triplet_id":"openstack%2Fpuppet-barbican~master~I8a42a4377b1d06aae25eef6324c4a1b59c45773b","project":"openstack/puppet-barbican","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-22 08:21:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a42a4377b1d06aae25eef6324c4a1b59c45773b","subject":"Drop deprecated api_config","status":"MERGED","created":"2025-04-17 14:40:44.000000000","updated":"2025-04-22 08:22:56.000000000","submitted":"2025-04-22 08:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947591","meta_rev_id":"4700aec3845444ee37c2b3a3c46456dd8ebc6b88","_number":947591,"virtual_id_number":947591,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 08:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-04-21 14:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-04-22 07:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-04-22 07:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-17 16:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-21 14:11:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-04-22 07:16:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7ff6d1afc56552f52bd06a00dd23d1a220afe1b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-17 14:40:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1816ec8d7e2afe356f39122f78f374f11e636df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 16:05:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2880ef8df65844f4a688b8d7dda64596\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7a7cffd739845fab02af953e640d504 : SUCCESS in 2m 17s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7169cdb1db7b4c75b49a36df0cc570b9 : SUCCESS in 2m 17s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79afbe2653204c37b54938f232d9c9c5 : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/947a12284f5c4fc4a3dca263fbb294c4 : SUCCESS in 6m 49s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0f3834f219a461f962ae1753beb7e3f : SUCCESS in 4m 11s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dd458a042464f3999fa875adb598685 : SUCCESS in 3m 40s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d582fa3936d4e27ad5ae1d9db88c722 : SUCCESS in 30m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ae12803d85b4e1a8aafb57c466413df : SUCCESS in 1h 17m 14s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e05a031fb5144fb99038c84cf739a5b6 : SUCCESS in 8m 03s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69b8b8e1b2e54020b86839ebbdf9153a : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bf71b02dfd844f9b4887d332f058db8 : SUCCESS in 8m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e56ae418dd7e4be48327a6806de23f16 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8df7d723adff02fa06b98a615e29a02d1e468709","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-04-21 14:11:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4609997c64b3db8c539096beabe6d5710f15999a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-22 07:16:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fd327fad373578223e198e7923b024ba3dd2a26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 07:17:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"525657bbb6d3aa8985cbe225887017b82be3dfa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:21:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3360dd9b10204335bd9dbab0dfe151ee\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f446b1ebf3948a9bc9b76e1505907fd : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b895c4053e94bdb9bef8787d21aff06 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0c53f3cee914341807d5ce735e91ac7 : SUCCESS in 2m 33s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/76bd7251e0db4d6aba3bf75ae118f3b7 : SUCCESS in 7m 48s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2455732a64c4050aa016019c10e4cc6 : SUCCESS in 3m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6fd7a0d4a8042d09382d004e73e77ba : SUCCESS in 1h 03m 37s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e06a4bd4ec64c2883e4900349c2be34 : SUCCESS in 15m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b0306bff6d24f1d8e08a24444118ab4 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c7ef83c95962df3055b019a50cdec8339c4ceb51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:21:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4700aec3845444ee37c2b3a3c46456dd8ebc6b88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:22:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009ff25020694da29c6cb89b3d708459\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34e9b010b65841bc88e06766b597af82 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"073977f158098138a06f754bdbe6a23d238481d6","revisions":{"073977f158098138a06f754bdbe6a23d238481d6":{"kind":"REWORK","_number":1,"created":"2025-04-17 14:40:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/947591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/91/947591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/91/947591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/91/947591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/91/947591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/91/947591/1"}}},"commit":{"parents":[{"commit":"93ed4b6874bf5f326e8cc3a26794805da1499d8f","subject":"Remove watch_log_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/93ed4b6874bf5f326e8cc3a26794805da1499d8f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-17 14:40:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-17 14:40:09.000000000","tz":540},"subject":"Drop deprecated api_config","message":"Drop deprecated api_config\n\nIt was deprecated in favor of the barbican_config parameter[1].\n\n[1] 0a1c988b0f4344cb8fc7872d605276475960c67e\n\nChange-Id: I8a42a4377b1d06aae25eef6324c4a1b59c45773b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/073977f158098138a06f754bdbe6a23d238481d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/073977f158098138a06f754bdbe6a23d238481d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
