)]}'
{"id":"openstack%2Fpuppet-barbican~962350","triplet_id":"openstack%2Fpuppet-barbican~master~I2467346e77ecb853f7b63cc072a4bfca1e0a4362","project":"openstack/puppet-barbican","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-30 20:10:07.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-09-30 20:10:07.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-09-30 06:57:54.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I2467346e77ecb853f7b63cc072a4bfca1e0a4362","subject":"Refactor api service management","status":"MERGED","created":"2025-09-26 13:23:54.000000000","updated":"2025-09-30 20:10:07.000000000","submitted":"2025-09-30 20:10:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962350","meta_rev_id":"dcdc3717f29daebae8a40bf10d6ac06027f77a10","_number":962350,"virtual_id_number":962350,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-30 20:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-09-30 06:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2025-09-30 18:37:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 17:00:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 06:16:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-09-29 14:18:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a85de06e6ad93e34df37e323ce157085b2a40f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 13:23:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"411c8a7800b8085e95d9bf556f9985af8b72a713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 13:25:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5a725a5d56d903ec7cba77e78ec23aeb6e0e729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:14:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"10a7e74121f47e45154ca4f30bf8e3759619ae98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:42:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d3f6caf0b7fd16afb5c405f8017359f5b22a4d07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 15:29:28.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"eaab0f94e15dc10f9f7db51bf9dc917284f93e5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 15:34:12.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"ab989bb62b3d34a21579f6c73d0023b8cbe1a8ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 17:00:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/923ccea595d54e1cb4295488d81fab3b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9aa56eb565764e6c9fb68e9416522821 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/882e0d057a114efc9da48aeecb371876 : SUCCESS in 2m 20s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/8c82d481c3404ed188af50b66aa9a008 : FAILURE in 6m 53s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05c5841eb7154ffba991c987521498c9 : FAILURE in 4m 55s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5f2fdea88964a4aaef91b1a8021e92e : FAILURE in 3m 03s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b659e07851346758dbc7931c87520bf : SUCCESS in 31m 06s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f058904d70a4918a1538e924671267b : SUCCESS in 49m 48s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b7e2fcef146437b85cf9a651ba569c9 : SUCCESS in 1h 20m 10s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b3dc79e278c4541aba8928394983eb1 : SUCCESS in 17m 18s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8488736b1c247cabd8cb9e51800db26 : SUCCESS in 7m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7047bea809474ee9acb7c7ec3b2b8d70 : SUCCESS in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8de08012be4017410b6d0fe8bbd21a8b66ddab39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 22:11:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"29880e219f61276d7f5a62270aa8f361ef5a216f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 23:37:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05a0e93dc0474296881c7f56f61c70e1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b88775594b447d282acd2231b4f268b : SUCCESS in 3m 37s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d608d15648434589a8fd25c0e0c438ed : SUCCESS in 2m 53s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/d9fe5a7c32e14ca8a71d041754ac0441 : FAILURE in 3m 10s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d77b700d2e0b434890b15a9267b71fed : FAILURE in 2m 57s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1c56dbbc326465485a4302a8270e128 : FAILURE in 2m 23s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83a1b9e090ec4061827622915186d557 : SUCCESS in 59m 04s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a18d69ccc7134829be8b1e97670dfd3b : SUCCESS in 1h 16m 37s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1be09f22182d477c8e207a83aeed1676 : SUCCESS in 58m 33s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4227cc4d1694796a8e0bd15fff9b772 : SUCCESS in 9m 00s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4147360b0c33413abb856df941c7b394 : SUCCESS in 8m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20dec128b62c487c81b0c054650d7ea0 : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a230c6c115832313b217aa7c561c50b500ad7cd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-27 03:12:23.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"cc15cfb6ae91ce7be194e94e42a2eb8583b46ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 04:28:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b076e967b9d34975983e6c4811d2a15f\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b838b712b024182b639efb781e09c46 : SUCCESS in 2m 19s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe0e6e61252c46fe85a595af2dae364d : SUCCESS in 3m 38s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/2c8e606986cb47a48fb04c2b9e160e4a : FAILURE in 3m 43s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fadc650d0b004e9db5b874cfb644f5dc : FAILURE in 3m 34s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28e8b91d2f88447fb285381d39cc4e64 : FAILURE in 4m 14s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9de3760df7334ba99722c2d4a3b88a58 : SUCCESS in 1h 05m 10s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0222b9e81554d18bde36a5f7798060c : SUCCESS in 1h 07m 55s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6494eae4c671461e9788703315b11617 : SUCCESS in 47m 39s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d22022a1e2c463f9c23c067d970b893 : SUCCESS in 13m 02s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1692f0c2dc84065929499364d6dee10 : SUCCESS in 10m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/945e7dbda086497cac8cd5a9dd53ddac : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"556c20f2311c55d90b4c25c9f987713f9ef22b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-27 05:50:14.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"43cf38ad3d7878d191d134219790a5c3313aabef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 07:12:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90078b1c94a642eeb445ad18cc13981a\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d9831362e6040f8b1694b02feb9d80d : SUCCESS in 2m 54s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddf1ee67923a4743bb42c424a6b5e5f2 : SUCCESS in 2m 42s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e6ce0738b5794920bf98a4486ddda402 : SUCCESS in 3m 28s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e7371bc5d6b49e68409420339bfb2f9 : SUCCESS in 3m 51s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32505b1cdded4bd8a902a2fdc70b4d0f : SUCCESS in 4m 34s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da46dc8f2f56428eac4e4ce643a062b2 : SUCCESS in 1h 02m 08s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6b7ff9723c74b88a41ef90af0fc85d1 : SUCCESS in 1h 17m 26s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a80cff07ae7547ebb5792f882ac9005b : SUCCESS in 1h 17m 44s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2ae6ea9608942b793de9076549e53be : SUCCESS in 15m 54s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7661760a78ba4014a04c0e94b0c78e9a : SUCCESS in 6m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02c9f22a5fab4ca6ac47dd26b7aac310 : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d14a95643969cd71d88e8788cc32d0e2a9befa27","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-29 06:16:58.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"39200184b8d285cdd07b66a4ce50580e377874f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-29 09:07:06.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"87bc3a8f04920e5d45ed9e7a38bc5879cae20fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:35:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f60a6a72e36644ef8fc40bf7f4292531\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e4a1db7f0314c72972ff7c2a21daa47 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd262c849b1949b286a005c4a0e9ac6e : SUCCESS in 1m 44s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/2b63bdfc0de64faa97f0dc0a46d9e4ef : SUCCESS in 3m 31s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e0adbf84b7c4376927ac54530f20182 : SUCCESS in 4m 44s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/241ef5931b0640948ba7a16b298ec0c4 : SUCCESS in 2m 43s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/991e569a0c3f40efbf4f0baae4d114d7 : SUCCESS in 1h 04m 53s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa6d287549744ff6bc7a9ff35b6badd8 : SUCCESS in 1h 01m 32s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abf8456d94a4476ebd2898b33228393e : SUCCESS in 1h 23m 09s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e2efe91e6d04f79b8fdd9185592082c : SUCCESS in 12m 57s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ee1a68da6eb42f49eba1a8a0128d8ad : SUCCESS in 7m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/848bb53701c04d6e827bd7b1069c990a : SUCCESS in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c25c4b185aab8891816f085a53d30b342b9717e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-09-29 14:18:28.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"53043224df1af26056a8e631ab92b1770acbe9f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-30 06:57:28.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"84a7942829f6f7cc58cf48bf8cb7d335a2b5e9b2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-30 06:57:54.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"112228cbd86dbd2c269072124ea42da028a6dd22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 08:23:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/031ba905cfe8483eac1907bb873c7728\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/252ea6a4fe6148edbbe17f5c44482f49 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa3128fdf7b5435b82867a0ff868395a : SUCCESS in 2m 50s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/af1663149c83495bb1212bcde54e1306 : SUCCESS in 3m 37s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cf2c59833fb4c4fbcaed8fc7b517ed1 : SUCCESS in 3m 53s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c1ab871c23a4571a617a9ac92388383 : SUCCESS in 4m 28s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24e32692901e409688c7f1f8cb1c4978 : SUCCESS in 1h 03m 54s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2cfa826faa748ef8a21a0bb3940cb4b : SUCCESS in 1h 20m 59s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/296be41236eb476ba39f0ff0ea1ae4af : SUCCESS in 1h 17m 19s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f154830c622424b880be4d71a3e81c2 : SUCCESS in 8m 11s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b72c522a14464a8fa6984a3ab5f178b3 : SUCCESS in 11m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a8ae2748b00439abfa490eb4612f740 : SUCCESS in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"da4c8803f120e6b1a4a1a1e4d265b41c1e062141","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-30 18:37:14.000000000","message":"Patch Set 11: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4fbde3c765f0caa23fe2b23776c84f34e12fff73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 18:37:54.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1b68fe61903758ce05773937b980f4b5cc204772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 20:10:07.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/928f2e77c24e4247897f6824a3fba7a5\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbf0d47ea1b44792a79738c57d38aa1d : SUCCESS in 2m 25s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/903f69c86193486191a4e472f70b50ea : SUCCESS in 2m 53s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/7e37467d13b04b40af9a2b064c705b39 : SUCCESS in 4m 24s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/869aaa052dbb4541b8cb67ce0294a4ec : SUCCESS in 4m 01s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f34793bc03b4f4dab40815aff353e01 : SUCCESS in 1h 28m 12s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/67e2b109dd87403d9d5fd90a1e4a4ec8 : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":11},{"id":"dcdc3717f29daebae8a40bf10d6ac06027f77a10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 20:10:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"bfc722352a3301cc8eee4c80099521be7c7826de","revisions":{"ec3389c61905f1098b73e06680aad1b7493e2f9e":{"kind":"REWORK","_number":1,"created":"2025-09-26 13:23:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/1"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:23:46.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ec3389c61905f1098b73e06680aad1b7493e2f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ec3389c61905f1098b73e06680aad1b7493e2f9e"}]},"branch":"refs/heads/master"},"9287cb8728e643bbf839253f0b606d8d5c9757b3":{"kind":"REWORK","_number":2,"created":"2025-09-26 13:25:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/2"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:25:44.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/9287cb8728e643bbf839253f0b606d8d5c9757b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/9287cb8728e643bbf839253f0b606d8d5c9757b3"}]},"branch":"refs/heads/master"},"17939bbfd19339417392dd3be7167c16e9f151ba":{"kind":"REWORK","_number":3,"created":"2025-09-26 14:14:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/3"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:14:30.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/17939bbfd19339417392dd3be7167c16e9f151ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/17939bbfd19339417392dd3be7167c16e9f151ba"}]},"branch":"refs/heads/master"},"0cc0f8e9fe0de88f7f31937d461807855db1f443":{"kind":"REWORK","_number":4,"created":"2025-09-26 14:42:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/4"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:41:41.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0cc0f8e9fe0de88f7f31937d461807855db1f443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/0cc0f8e9fe0de88f7f31937d461807855db1f443"}]},"branch":"refs/heads/master"},"ba6587b95c3b9ac5ec64de2a04bce6e528286673":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-09-26 15:29:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/5"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 15:29:28.000000000","tz":0},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ba6587b95c3b9ac5ec64de2a04bce6e528286673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ba6587b95c3b9ac5ec64de2a04bce6e528286673"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ced0a4f21a1c5f6d0d228c671af4e846c3de49b6":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-09-26 15:34:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/6"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 15:34:12.000000000","tz":0},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ced0a4f21a1c5f6d0d228c671af4e846c3de49b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/ced0a4f21a1c5f6d0d228c671af4e846c3de49b6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"998cda3f2860b88cc9aff333589afed48ab057a5":{"kind":"REWORK","_number":7,"created":"2025-09-26 22:11:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/7"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 22:11:36.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/998cda3f2860b88cc9aff333589afed48ab057a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/998cda3f2860b88cc9aff333589afed48ab057a5"}]},"branch":"refs/heads/master"},"564ecad0f5ec780b80c74e0d81482dad2e262ea9":{"kind":"REWORK","_number":8,"created":"2025-09-27 03:12:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/8"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-27 03:12:15.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/564ecad0f5ec780b80c74e0d81482dad2e262ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/564ecad0f5ec780b80c74e0d81482dad2e262ea9"}]},"branch":"refs/heads/master"},"57931c75a7069838287c6f783974ee7a3fff434c":{"kind":"REWORK","_number":9,"created":"2025-09-27 05:50:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/9"}}},"commit":{"parents":[{"commit":"1ad57ce5b0cb1a92591c88936ed116e6a0835e8c","subject":"Merge \"Import base parameter types for ::policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/1ad57ce5b0cb1a92591c88936ed116e6a0835e8c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-27 05:50:06.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/57931c75a7069838287c6f783974ee7a3fff434c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/57931c75a7069838287c6f783974ee7a3fff434c"}]},"branch":"refs/heads/master"},"7336ea02f503ec410898c588a88b29655470cb8f":{"kind":"REWORK","_number":10,"created":"2025-09-29 09:07:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/962350/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/10"}}},"commit":{"parents":[{"commit":"033369bc9c098aedad9d5cc3ce0ac743391c7458","subject":"Escape IPv6 address in gunicorn config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/033369bc9c098aedad9d5cc3ce0ac743391c7458"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-29 09:06:51.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7336ea02f503ec410898c588a88b29655470cb8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7336ea02f503ec410898c588a88b29655470cb8f"}]},"branch":"refs/heads/master"},"bfc722352a3301cc8eee4c80099521be7c7826de":{"kind":"REWORK","_number":11,"created":"2025-09-30 06:57:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/50/962350/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/50/962350/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/50/962350/11"}}},"commit":{"parents":[{"commit":"033369bc9c098aedad9d5cc3ce0ac743391c7458","subject":"Escape IPv6 address in gunicorn config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/033369bc9c098aedad9d5cc3ce0ac743391c7458"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 13:22:01.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-09-30 06:57:28.000000000","tz":0},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I2467346e77ecb853f7b63cc072a4bfca1e0a4362\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/bfc722352a3301cc8eee4c80099521be7c7826de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/bfc722352a3301cc8eee4c80099521be7c7826de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
