)]}'
{"id":"openstack%2Fpuppet-barbican~980313","triplet_id":"openstack%2Fpuppet-barbican~master~Ibf32775820f0f724e40c7c44e63710b38b440ce5","project":"openstack/puppet-barbican","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-16 23:43:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf32775820f0f724e40c7c44e63710b38b440ce5","subject":"simple_crypto: Require kek","status":"MERGED","created":"2026-03-12 15:47:41.000000000","updated":"2026-03-16 23:44:30.000000000","submitted":"2026-03-16 23:43:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980313","meta_rev_id":"306c4d7950f6c844eee42a25b7d6fc9c226251c5","_number":980313,"virtual_id_number":980313,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-16 23:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2026-03-16 14:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2026-03-16 22:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2026-03-16 22:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 17:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-16 14:10:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2026-03-16 22:13:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9d17f61eafd3874b6eb922e36af803430eaf217f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-12 15:47:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ff916cc5801c687d7088df0c54b9ae63ca71569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 17:10:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de1f2ba1678e4250905c8a72c91d0179\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06a461de1d404e51a97233d3c13ae9f9 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e19ad1b0be0430a915e8c2d45b49102 : SUCCESS in 3m 20s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/8684147a2fe747fe914acbbb0f409664 : FAILURE in 5m 30s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8a52338d84e48c8a7833140e7e00994 : FAILURE in 5m 55s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96872fb1808f4f3687c4915859fa40b3 : FAILURE in 6m 15s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e439241b8d574d04b98367ee3e83a816 : FAILURE in 1h 17m 21s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58e6a333eb6f4bd38a8ee8594cf13c0a : SUCCESS in 1h 13m 11s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72ecc395bed147fe817f8b0c0b1bc21d : SUCCESS in 42m 29s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6143fd819024a6dad6574ea3630083d : SUCCESS in 20m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff148ebb4c4642cfb29a91477e0507a9 : SUCCESS in 8m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/098799ee107c439da53370edfe7e2aad : SUCCESS in 13m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f53978a9c704136b00aa8efb8a2d5d3 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"12f83daeaab537760d4d8898b66376f2d7d6f957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-13 12:56:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"69c978e7df4003c0227430d9a080c3b07cef5106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 14:22:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b04d5b3d8f6d411481ebd6e70e83cc70\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13d1b8d980154973a07ffa579b1fee9e : SUCCESS in 2m 59s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e440084e87bd40bdbebb5a09c81bb367 : SUCCESS in 1m 44s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/d4b9a533c38b43809d2a48e4fdca5cbb : SUCCESS in 7m 13s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d293c1056554972896bd4eb5b82904c : SUCCESS in 5m 39s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b386c6ce6a3947a4b5ca2d9487303c92 : SUCCESS in 5m 03s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb4f1d083e344fa2a3c839efdf2ebcc1 : FAILURE in 1h 13m 31s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/89a71825fb5c48579a1bc481a9e28cdc : SUCCESS in 1h 10m 06s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c471dfa383df4701bfae9fcac85e2af2 : SUCCESS in 1h 20m 33s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d99a3be408854344ba0cd67bb99a6fad : SUCCESS in 18m 23s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98dff84b72a24f6d83e2a14d8d34db48 : SUCCESS in 12m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b12a0fbc9cb54998b0f50cbe452beec0 : SUCCESS in 13m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d0c4b141ae46adaa7910a35c0e7bc5 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":2},{"id":"124aff19ee5d57afc28d7893467da8d2cda148d3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-03-16 14:10:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"be418199bee967cad4316d9cb7f9bd66c6ef97ca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-03-16 22:13:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fc92b0cfebe8ac2f8bae131df4b83eb15ecf53a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 22:14:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf66b87d67f99932e3e9ccf8eb313ee6b5c3ce4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:43:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c43812bba714811a6bbb3b50fd064db\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33c55f608fc449f0a80c92af01a5de7d : SUCCESS in 3m 35s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c22a47cb1f34dcf947751398576c8f7 : SUCCESS in 3m 10s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/3f0c635fa8544c1a840af559e76a58d6 : SUCCESS in 5m 34s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/022542fd9b52410e8e273948faa62b19 : SUCCESS in 4m 39s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/131b30554d944ed399eaa037034c8312 : SUCCESS in 1h 23m 58s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebee90d876c2430584d0b30283598ba8 : SUCCESS in 17m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fa491ac6f99474eb839903a61a93363 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a9c950af23b1287937051d5727950ea80797db53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:43:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"306c4d7950f6c844eee42a25b7d6fc9c226251c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:44:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d637af1577e4aa5b64fe2b8a0b75160\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbbc9662c5824f498c7988c5da91b807 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d980cfd710fc976d1152498c4944303519f547d8","revisions":{"7a432b223367eadf5788892a78337a9c249cf6a9":{"kind":"REWORK","_number":1,"created":"2026-03-12 15:47:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/980313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/13/980313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/1"}}},"commit":{"parents":[{"commit":"a0faf98f469fe6409fab06c5c3dff4377ca55209","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a0faf98f469fe6409fab06c5c3dff4377ca55209"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-12 15:46:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-12 15:47:28.000000000","tz":540},"subject":"simple_crypto: Require kek","message":"simple_crypto: Require kek\n\nThe option is required by barbican since barbican 21.0.0[1], without\nany deprecation phase. Require the parameter so that users can notice\nthis requirement early.\n\n[1] 5250c5f368204083fd36bbed5273b61ebe9e6de4\n\nChange-Id: Ibf32775820f0f724e40c7c44e63710b38b440ce5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7a432b223367eadf5788892a78337a9c249cf6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/7a432b223367eadf5788892a78337a9c249cf6a9"}]},"branch":"refs/heads/master"},"d980cfd710fc976d1152498c4944303519f547d8":{"kind":"REWORK","_number":2,"created":"2026-03-13 12:56:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/980313/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-barbican","ref":"refs/changes/13/980313/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-barbican refs/changes/13/980313/2"}}},"commit":{"parents":[{"commit":"a0faf98f469fe6409fab06c5c3dff4377ca55209","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/a0faf98f469fe6409fab06c5c3dff4377ca55209"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-12 15:46:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-13 12:56:13.000000000","tz":540},"subject":"simple_crypto: Require kek","message":"simple_crypto: Require kek\n\nThe option is required by barbican since barbican 21.0.0[1], without\nany deprecation phase. Require the parameter so that users can notice\nthis requirement early.\n\n[1] 5250c5f368204083fd36bbed5273b61ebe9e6de4\n\nChange-Id: Ibf32775820f0f724e40c7c44e63710b38b440ce5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d980cfd710fc976d1152498c4944303519f547d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-barbican/commit/d980cfd710fc976d1152498c4944303519f547d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
