)]}'
{"id":"openstack%2Fpuppet-ceilometer~773212","triplet_id":"openstack%2Fpuppet-ceilometer~stable%2Ftrain~Ibadafdf86f5ce581c9084a5e00784eefa7d28cab","project":"openstack/puppet-ceilometer","branch":"stable/train","topic":"bug/1912345-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"Ibadafdf86f5ce581c9084a5e00784eefa7d28cab","subject":"Remove transformers from pipeline.yaml","status":"MERGED","created":"2021-02-01 13:27:06.000000000","updated":"2021-02-23 18:50:54.000000000","submitted":"2021-02-23 18:48:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773212-bug/1912345-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":772664,"cherry_pick_of_patch_set":1,"meta_rev_id":"71eb4dd039a378faff0394dd085a33a5955c9233","_number":773212,"virtual_id_number":773212,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 18:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2021-02-02 16:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-02-10 14:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-02-23 17:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-02-23 17:48:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 14:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 16:48:24.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2021-02-10 14:25:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-02-23 17:48:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a578247fc3b84b96c1ea7a1e36321d186c47a55d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-01 13:27:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"6dd56d07104f5a8075278ffb44cf50f2448de3f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 14:30:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72bcfedf39a54c87b19f6265fb9b1d92 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed7b2a8147254aab9e1f79c460b97f63 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5a44571031342ce973bc5a79f37e3b9 : SUCCESS in 4m 56s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/87835f24e4cc4e5ab43c256e82249d82 : SUCCESS in 14m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4ae437859fa648eaa7ec4215d2f6a0b0 : SUCCESS in 12m 53s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/dfcf6287582f477fa873e84f9bb7244e : SUCCESS in 8m 49s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1a30ce32ad14412aa717398d6791acc : SUCCESS in 10m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3dfd5142f284d4da59b984d0d9deff5 : SUCCESS in 9m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a2255357bd8345d7bb4fe68ee90c1a8e : SUCCESS in 47m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3ee9bccd0da74a8c8bed5d45ed59d6f8 : SUCCESS in 52m 09s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2a490f9931f94e119b10d51c3ee9c3db : SUCCESS in 49m 11s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b9a9a7ca6b02408190a1344b78b38335 : SUCCESS in 17m 46s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78ab2ca2da4e4027916ec63ce8b72879 : FAILURE in 5m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dc0933f58f544e1968284db1d90a15d : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"43e6f6cf903f2f48448530004e7a709bd1807793","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-02-02 16:48:24.000000000","message":"Patch Set 1: Code-Review+1\n\nthank you, yes! this makes sense.","accounts_in_message":[],"_revision_number":1},{"id":"2afb2b38e1e562d4fba0bcdb8fef890cf37a1da1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-10 14:25:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c87cb91c97a21937e9eb23b0975afeb8a517353a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 17:48:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9838e0e17e4488b2594fc19f8dd4e8e86c865601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:48:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85d7b1cc2fc61455cdaa45fa41b8afd987c0e416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:48:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e018bc1820ec4c0cbaabe94f10ec5cc7 : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87c2ac57cf9d4cae8f91da97fadbe1be : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7f2ef1de4fe4f159f780411bb460ad8 : SUCCESS in 6m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/95b3c2ad95124c1c945be06ec6305c46 : SUCCESS in 17m 49s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83d150cc8fb34c40a42f2090dde65d3b : SUCCESS in 12m 18s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6ad88f71992846dda50ccd7c5936077a : FAILURE in 31m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/d618e3a92946457bbae961eac13cfe26 : SUCCESS in 51m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/466af1d2bfe44eb980946651f580be8c : SUCCESS in 15m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ee9626234f4faf9b4391ef90cd4090 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4eba6156ec68d182d7f333c8b6f2c68f132fe306","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:48:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"71eb4dd039a378faff0394dd085a33a5955c9233","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b42361af8c04a589a305e5c04cc6013 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5901c94b2cb90ec51b722e877be46edbdc80dba0","revisions":{"5901c94b2cb90ec51b722e877be46edbdc80dba0":{"kind":"REWORK","_number":1,"created":"2021-02-01 13:27:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/773212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ceilometer","ref":"refs/changes/12/773212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ceilometer refs/changes/12/773212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ceilometer refs/changes/12/773212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ceilometer refs/changes/12/773212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ceilometer refs/changes/12/773212/1"}}},"commit":{"parents":[{"commit":"d502884411343c862cf056fc2bf25ec6342548dc","subject":"Replace disk.* meters by disk.device.* meters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceilometer/commit/d502884411343c862cf056fc2bf25ec6342548dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-19 12:56:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-01 13:27:06.000000000","tz":0},"subject":"Remove transformers from pipeline.yaml","message":"Remove transformers from pipeline.yaml\n\nThis change removes usage of transformers from pipeline.yaml because\nsupport of transofermer was already removed from ceilometer[1].\n\n[1] 9db5c6c9bfc66018aeb78c4a262e1bfa9b326798\n\nBackport note:\nThis backport patch includes the subsequent change[2] to fix the wrong\nfile mentioned in the release note file.\n\n[2] d27fbd3fed472a51670393a4a295ba6b16a35bea\n\nCloses-Bug: #1912345\nChange-Id: Ibadafdf86f5ce581c9084a5e00784eefa7d28cab\n(cherry picked from commit 5b8f5d0490f43f43a62cee9718c75290e30473cf)\n(cherry picked from commit 815d7033dbe6f5c3dde1128ce6aecc4048eb09cb)\n(cherry picked from commit ed9563b63fdc85f16b95f9daee311f26e21a391e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceilometer/commit/5901c94b2cb90ec51b722e877be46edbdc80dba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ceilometer/commit/5901c94b2cb90ec51b722e877be46edbdc80dba0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
